)]}'
{"id":"openstack%2Fplacement~677504","triplet_id":"openstack%2Fplacement~master~I81d36e7d0d082423a4c8e31e9a01f5e0285431be","project":"openstack/placement","branch":"master","topic":"api-ref-aggregates-typo","hashtags":[],"change_id":"I81d36e7d0d082423a4c8e31e9a01f5e0285431be","subject":"api-ref: fix typo in aggregates note","status":"MERGED","created":"2019-08-20 16:17:09.000000000","updated":"2019-08-20 18:42:27.000000000","submitted":"2019-08-20 18:37:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677504-1566326227042-3b288848","meta_rev_id":"559ce2c1ce24010e4b8d21ca36c95f7b383b76b9","_number":677504,"virtual_id_number":677504,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-08-20 18:37:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":2,"date":"2019-08-20 16:17:55.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":1,"date":"2019-08-20 16:17:55.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-20 16:17:55.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-08-20 18:37:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c56ad7586f2e78c1709f0fb827115edbf9c6c03","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-08-20 16:17:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dbbc8dc3d78f593cfa849f94d0b1c18f9fdd91a3","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-20 16:17:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nyou stat padder, you","accounts_in_message":[],"_revision_number":1},{"id":"684fc3f2ad17c9547ef057294d074c06a99abf16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 16:44:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0a19ca05405e42a5a50f1757c7504dd2 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b710374f53594a8895ffc507b6f10a88 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55adc332bd8b41f681461b15a9c62441 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":1},{"id":"e52f9930fd33c13df58d67c804b9a160ecbbcee5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 16:44:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca91eb1fc2be095a9589aa5446c4199586ec2c87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 18:37:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1c3896c11ea840ae92cd78c39327a8a3 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9790dcde82241fdaca8a065d138f71b : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afe4903c15f14d279fe72ffeef84635e : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":1},{"id":"d0800f2afe1a395dbdc608065244eae272fb6d6e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 18:37:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"35523cacfe00fae768fafc5448948bbf57c77600","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 18:39:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/37e300b9625242dfbf4048106e4653db : SUCCESS in 1m 47s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3a57e931a384967bf663e5be83214ce : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1},{"id":"adaec1c9f62dc3a492c2e020978df9af38ef5b35","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-08-20 18:42:27.000000000","message":"Patch Set 1:\n\n\u003e you stat padder, you\n\nI actually noticed this while reading the API reference while reviewing this https://review.opendev.org/#/c/640898/ I swear to all that is holy.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"31794c854685ed28ee5037f8f668a04f2874af87","revisions":{"31794c854685ed28ee5037f8f668a04f2874af87":{"kind":"REWORK","_number":1,"created":"2019-08-20 16:17:09.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/04/677504/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/04/677504/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/04/677504/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/04/677504/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/04/677504/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/04/677504/1"}}},"commit":{"parents":[{"commit":"aecd0f3497c5e4f1e73bf6c76a9add3db6270e27","subject":"Avoid duplicate ProviderSummary in _merge_candidates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/aecd0f3497c5e4f1e73bf6c76a9add3db6270e27"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-08-20 16:16:54.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-08-20 16:16:54.000000000","tz":-240},"subject":"api-ref: fix typo in aggregates note","message":"api-ref: fix typo in aggregates note\n\nChange-Id: I81d36e7d0d082423a4c8e31e9a01f5e0285431be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/31794c854685ed28ee5037f8f668a04f2874af87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/31794c854685ed28ee5037f8f668a04f2874af87"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
