)]}'
{"id":"openstack%2Fplacement~679480","triplet_id":"openstack%2Fplacement~master~I76cca613e0d89e235f7fc23740dda93bda4ffd15","project":"openstack/placement","branch":"master","topic":"cd/yet-another-by-cdent","hashtags":[],"change_id":"I76cca613e0d89e235f7fc23740dda93bda4ffd15","subject":"Fix typo in microversion sequence test","status":"MERGED","created":"2019-08-30 16:34:22.000000000","updated":"2019-08-30 19:45:22.000000000","submitted":"2019-08-30 19:43:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679480-1567194181312-c9d86742","meta_rev_id":"173d4e89a322aba03bc41a2204e862c29d09829c","_number":679480,"virtual_id_number":679480,"owner":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-08-30 16:36:10.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":2,"date":"2019-08-30 19:43:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":2,"date":"2019-08-30 18:30:17.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":1,"date":"2019-08-30 18:30:17.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-30 18:30:17.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-08-30 19:43:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bff56bd14e9efd1d7fd509134c359cdee07f6e79","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-08-30 16:34:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9385a9089226782b3cb39337ff173a46b39ec0c","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-08-30 16:36:10.000000000","message":"Patch Set 1:\n\ncrikey, I made a typo in my topic for a typo fix\n\ncd/yet-another-by-cdent\n\nwas supposed to be\n\ncd/yet-another-typo-by-cdent\n\nthe meta, it kills me","accounts_in_message":[],"_revision_number":1},{"id":"251510948fb5948aeaa742f02383c754ae5cde32","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-30 18:30:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a96a9d9c0a3813e82ba43abc9348027e9bfbc439","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 18:59:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7c47993377e45b3b91754c43d158bac : SUCCESS in 5m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89a1e5cd979c4ab18d6ae9035a005ff1 : SUCCESS in 5m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d351058d8644eb79bd2a3e9cf7c29a4 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b345dac301704ca9a47e185f3c875d08 : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/308c7973251e4980ab230c0c69f661f0 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae145ece5701414397c8a46ead28711b : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9f5ac0a31b2941cd9696198aabc80321 : SUCCESS in 6m 17s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/bf97d1f5980f4d018c95341024d4a05e : SUCCESS in 5m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0b054455a5484595a09273c27c5ba35f : SUCCESS in 5m 01s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/738826d9820e47209c0002ab5164fc6c : SUCCESS in 5m 15s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/cb8e4b06f5e34457af391e7e501c651c : SUCCESS in 12m 46s","accounts_in_message":[],"_revision_number":1},{"id":"2538b421b62bad3f27ab589ac6a7d844f09fadca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 18:59:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"779a1c8fc593b25039a341eed0b7d67c565f3496","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 19:43:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/431869ff79eb4e48adeb51779cd285d2 : SUCCESS in 7m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14fe3056a31b4d88b8c6b31c571ae816 : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0794ccec0fe40e588bceee1b7f2e320 : SUCCESS in 7m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5759474cdc444dd78f7f66cec4add522 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/873a49a94b9b4b9a824e610178c2dd73 : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eb5f70f8e48f42fda1c1c622ded38f51 : SUCCESS in 5m 25s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8fe88f8bafef41018772641360a629de : SUCCESS in 8m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0114ac47e391467291a89393689c08d3 : SUCCESS in 5m 22s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/d9a785bf270245729bcfc64eb005c6c7 : SUCCESS in 5m 53s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/923321049af148b7b8db50fc94514710 : SUCCESS in 14m 48s","accounts_in_message":[],"_revision_number":1},{"id":"a74be99203aab5a8aeb7fb762684c67df4f2a8b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 19:43:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"416fa8ef1f368ee4eac4f8a878c8d1f8370ae7a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 19:45:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7a4780148a14355ad3487b54f25ffcd : SUCCESS in 1m 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"156b08335733f6b13bc994c831bdf5b7add7f417","revisions":{"156b08335733f6b13bc994c831bdf5b7add7f417":{"kind":"REWORK","_number":1,"created":"2019-08-30 16:34:22.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/80/679480/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/80/679480/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/80/679480/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/80/679480/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/80/679480/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/80/679480/1"}}},"commit":{"parents":[{"commit":"8272dfee98d800566b4e2d169d0fa02a635293c1","subject":"Fix links to migration scripts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/8272dfee98d800566b4e2d169d0fa02a635293c1"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-08-30 16:32:55.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-08-30 16:32:55.000000000","tz":60},"subject":"Fix typo in microversion sequence test","message":"Fix typo in microversion sequence test\n\nThe assertion failure message in a microversion sequence test\nhad a typo. This was not noticed because the test usually\npasses and the message is not produced under normal circumstances.\n\nChange-Id: I76cca613e0d89e235f7fc23740dda93bda4ffd15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/156b08335733f6b13bc994c831bdf5b7add7f417"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/156b08335733f6b13bc994c831bdf5b7add7f417"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
