)]}'
{"id":"openstack%2Fplacement~681047","triplet_id":"openstack%2Fplacement~master~Ifb14d2c14b17fc5bcdf7d124744ac2e1b58fd063","project":"openstack/placement","branch":"master","topic":"story/2005171","hashtags":[],"change_id":"Ifb14d2c14b17fc5bcdf7d124744ac2e1b58fd063","subject":"Deprecate [placement]/policy_file config option","status":"MERGED","created":"2019-09-09 15:22:12.000000000","updated":"2019-09-26 17:31:21.000000000","submitted":"2019-09-24 14:24:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681047-1569335081563-fe8b4c94","meta_rev_id":"3d438be5d0bf5235f9da7381146a08324bbcab67","_number":681047,"virtual_id_number":681047,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"date":"2019-09-24 13:35:09.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-24 14:24:41.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2019-09-24 12:34:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-09-23 10:53:13.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2019-09-24 12:34:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-09-13 08:30:14.000000000","_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-09 15:22:47.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2019-09-09 15:34:41.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-09-10 15:27:46.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-09-13 08:30:14.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2019-09-23 10:53:13.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-09-24 12:34:36.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-09-24 14:24:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eef9fcf1e0844a7f3297e503aa441f999565dda3","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-09 15:22:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9fe649d2e5ed5d105eb67d99a138ad6740e5f60","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-09-09 16:39:41.000000000","message":"Patch Set 1:\n\n(3 comments)\n\nOverall this looks reasonable to me. I don\u0027t know that we have a cleaner way to handle this type of situation.","accounts_in_message":[],"_revision_number":1},{"id":"b68a3166ca12df06b2036f5660a53e426e6c5556","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-09 16:59:42.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"db701e44e7614a5c5db14fd3e7933fe025e87483","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 20:47:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3f63c827e53b4bbebbb78b950d54279e : SUCCESS in 1h 00m 01s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/650c43c0d0d84b1eaa4ebe0bf327c379 : SUCCESS in 1h 18m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64b4518e77b34942aa68dac449882613 : SUCCESS in 4m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2182a47d4594df0a0ddb937c31f12ac : SUCCESS in 7m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20ed9bf7d8bc40e5a1411b8b20b9c3d9 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af7905163f844648ad7d380edf96bbaa : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7557953827934574892a824185df91a7 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d643ed1445dd4a06828eb9bd35c78789 : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8a2b519f3e93418b883c19dfd770928b : SUCCESS in 6m 54s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/607944939b17458087a6496b05a65364 : SUCCESS in 5m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ce3912a8a94041da9aaefafa6bc12b83 : SUCCESS in 5m 11s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/3ad86abcfd4a46bd8d70a1cfe985370f : SUCCESS in 5m 13s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/17dbf1d81a644ac8b29ace238f731e29 : SUCCESS in 13m 12s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/5ad99e6210c14cc3b724d62ecd74727d : SUCCESS in 30m 35s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/d42b9e4201a54f72a074a4158d8c42cf : SUCCESS in 13m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/52dcb4bb36b94ce3a45f0a6e2734820d : SUCCESS in 1h 06m 44s","accounts_in_message":[],"_revision_number":1},{"id":"38ad6c1d62d4c3f73e99ce11f78d943faeeaa43e","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-10 12:42:15.000000000","message":"Patch Set 1:\n\n(1 comment)\n\napart from the TODOs seems reasonable.","accounts_in_message":[],"_revision_number":1},{"id":"e321072f9f32975fe97289628d163e34ef99c6e8","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-09-10 14:18:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2586a92f6acddfe34121f45ebdf9e00f6e3ba4a6","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-10 14:51:53.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1a3b46993de0407b7ea3eff47e6ff479e6c9012c","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-09-10 15:27:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5f29da372c1d29f7f78b6523fd6dec94f65ac3ae","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-11 23:57:34.000000000","message":"Patch Set 1: Workflow-1\n\nI plan to update this tomorrow since it sounds like it\u0027s not crazy. I likely won\u0027t have time to tinker with it in devstack though until Friday or early next week.","accounts_in_message":[],"_revision_number":1},{"id":"0423478bb543c4efd968295c25f98f5d85e61496","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-12 22:19:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"27f93f0d9ab768e9a8d86e7af99050547e5c6773","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 05:56:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fba6209ac82e4e399f10f8ac34a0082e : SUCCESS in 58m 03s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/8275d565ce3c401ebad5cb606bac71da : SUCCESS in 1h 27m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/856e8f58d7ae4955af4c6b3c930f3c6b : SUCCESS in 4m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13283c3952c34b6a8df00648183e9f55 : SUCCESS in 5m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/253225dbdd07413bb91b0fcad96b3616 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c4b0fe62bf44cacaa75beb28e83049a : SUCCESS in 5m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6802d6b8d58f49ac83e9d746a23d9f88 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/063f7cea6a9c4052a03851902b25ede8 : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/54a8e6e451014c18bc2ca8f6490ab06b : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33385b23cab04d6b9a28a067a51b9d23 : SUCCESS in 4m 19s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/40cc062893984592aecd530265ac762c : SUCCESS in 4m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bed7a68a60b14bdbad33566022cf0e32 : SUCCESS in 4m 59s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/c410569b63b34bc48b9b35516a12ae7a : SUCCESS in 6m 48s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/16f98bbe18614a5eaf67d7e7c68c986f : SUCCESS in 12m 48s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/eb3c3a17a7ca4a6baf4ca09e312379f7 : SUCCESS in 9m 30s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/a7cea9dbaaca44e681f53463fd3e387f : SUCCESS in 10m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46fa68bd55b34291874f0d49d9e4dd39 : SUCCESS in 1h 00m 18s","accounts_in_message":[],"_revision_number":2},{"id":"dc7071c315d00a9861742f049afda68e6f439eaf","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-23 10:53:13.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks for doing this.","accounts_in_message":[],"_revision_number":2},{"id":"fc8bee2d13b06d558599b7cd51a448967ce4f25b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-09-24 12:34:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(3 comments)\n\nThere are some nits but we could do a follow-up for those","accounts_in_message":[],"_revision_number":2},{"id":"f671209b3463605254935e9cba11f99d1ef586a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 12:34:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f967744709866aaf2c7968a3fe10f0ecff116ed4","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-24 13:35:09.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7006b3faa72be404f98f87263ff4d6247bd7fd34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 14:24:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d1382d1c2f4f0304d8fe79a92c9c31d81e9f2994","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 14:24:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1de56ad8e10d4f63831d28cb4f66a65d : SUCCESS in 1h 07m 34s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/125e6cb820d346eda109a8c45d712c39 : SUCCESS in 1h 41m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56f2f616f0704165a153a8f78b7461c5 : SUCCESS in 10m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05efad26501f427aaf3f334340bff84a : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ab432d82e7945f7868492d908186c50 : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/94229133238d4ec39eba11562bbead64 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9eaee9cdc26d4e3c8a60d02f6a3436e5 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/81f615804c474680afbf4f5157b12e23 : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/999eb1e347244a79987bdcb872d25c90 : SUCCESS in 4m 51s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b3e05cae15894117af0d1548ab285432 : SUCCESS in 5m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/acffa10b22454051848153f9187cb665 : SUCCESS in 6m 39s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/9054afa23b7c4d869273f1528c1ff169 : SUCCESS in 5m 02s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/eb1b774174bf44d287706b510d8c1d13 : SUCCESS in 12m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5644a5f8f5a546ce85663c2946abf11d : SUCCESS in 1h 20m 54s","accounts_in_message":[],"_revision_number":2},{"id":"2a27ab9efd8c6d0ab90d1c43fffc693de87191b7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 14:26:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01afa9d0a15244b7852bcf71ad887de0 : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3dce2561bbb4f6f8d5b0fcfb558d20f : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":2},{"id":"cb11be15f619f27b61809cd73e0b45a265f799cc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-09-24 14:45:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d3023a70ee5a4bec62dc1f2dfc2e9f4d4628578b","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-26 17:31:21.000000000","message":"Patch Set 2:\n\nI got around to doing some manual testing in devstack with this, details are in this paste:\n\nhttp://paste.openstack.org/show/779523/","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9ece3d35ab728560b33852f6e19e5f92938dc2c1","revisions":{"b3e009068c3e663e5814c4122fec0c2896be69fc":{"kind":"REWORK","_number":1,"created":"2019-09-09 15:22:12.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/47/681047/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/47/681047/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/47/681047/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/47/681047/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/47/681047/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/47/681047/1"}}},"commit":{"parents":[{"commit":"10146978ea2d7eace1ffdf6552eafd22360e37dd","subject":"Fix misspell word","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/10146978ea2d7eace1ffdf6552eafd22360e37dd"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-09-09 15:03:12.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-09-09 15:22:06.000000000","tz":-240},"subject":"WIP: Deprecate [placement]/policy_file config option","message":"WIP: Deprecate [placement]/policy_file config option\n\nTODO: tests, update docs and PolicyFixture\n\nThe [placement]/policy_file option was necessary when placement\nwas in nova since nova uses the standard [oslo_policy]/policy_file\noption for defining custom policy rules. Now that placement is\nextracted (+1 release) we can deprecate the placement-specific\noption and use the standard [oslo_policy]/policy_file option as well.\n\nThe tricky thing with this is both options define a default value\nbut those values are different, and neither need to exist or can\nexist but be empty and we\u0027ll use policy defaults from code. So some\nlogic is necessary for detecting which option we should pass to the\noslo.policy Enforcer object. We prefer to use [oslo_policy]/policy_file\nif it exists but will fallback to use [placement]/policy_file for\nbackward compatibility. We also check for a couple of edge cases to\ntry and detect misconfiguration and usage of the deprecated option.\n\nThe config generation docs are updated to include the [oslo_policy]\noptions as well as registering the options from that library for\nruntime code.\n\nChange-Id: Ifb14d2c14b17fc5bcdf7d124744ac2e1b58fd063\nStory: #2005171\nTask: #29913\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b3e009068c3e663e5814c4122fec0c2896be69fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b3e009068c3e663e5814c4122fec0c2896be69fc"}]},"branch":"refs/heads/master"},"9ece3d35ab728560b33852f6e19e5f92938dc2c1":{"kind":"REWORK","_number":2,"created":"2019-09-12 22:19:01.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/47/681047/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/47/681047/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/47/681047/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/47/681047/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/47/681047/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/47/681047/2"}}},"commit":{"parents":[{"commit":"10146978ea2d7eace1ffdf6552eafd22360e37dd","subject":"Fix misspell word","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/10146978ea2d7eace1ffdf6552eafd22360e37dd"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-09-09 15:03:12.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-09-12 22:18:49.000000000","tz":-240},"subject":"Deprecate [placement]/policy_file config option","message":"Deprecate [placement]/policy_file config option\n\nThe [placement]/policy_file option was necessary when placement\nwas in nova since nova uses the standard [oslo_policy]/policy_file\noption for defining custom policy rules. Now that placement is\nextracted (+1 release) we can deprecate the placement-specific\noption and use the standard [oslo_policy]/policy_file option as well.\n\nThe tricky thing with this is both options define a default value\nbut those values are different, and neither need to exist or can\nexist but be empty and we\u0027ll use policy defaults from code. So some\nlogic is necessary for detecting which option we should pass to the\noslo.policy Enforcer object. We prefer to use [oslo_policy]/policy_file\nif it exists but will fallback to use [placement]/policy_file for\nbackward compatibility. We also check for a couple of edge cases to\ntry and detect misconfiguration and usage of the deprecated option.\n\nThe config generation docs are updated to include the [oslo_policy]\noptions as well as registering the options from that library for\nruntime code.\n\nChange-Id: Ifb14d2c14b17fc5bcdf7d124744ac2e1b58fd063\nStory: #2005171\nTask: #29913\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9ece3d35ab728560b33852f6e19e5f92938dc2c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9ece3d35ab728560b33852f6e19e5f92938dc2c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
