)]}'
{"id":"openstack%2Fplacement~681458","triplet_id":"openstack%2Fplacement~master~Icf7c22bf9d1de6fb2a74a756c370930d4c00b0b9","project":"openstack/placement","branch":"master","topic":"pdf-docs","hashtags":[],"change_id":"Icf7c22bf9d1de6fb2a74a756c370930d4c00b0b9","subject":"Build pdf docs","status":"MERGED","created":"2019-09-11 12:34:28.000000000","updated":"2019-09-23 17:26:55.000000000","submitted":"2019-09-23 17:23:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681458-1569259431642-f5fe0389","meta_rev_id":"50129a8387d74cf0e93338e35d17295e277d7214","_number":681458,"virtual_id_number":681458,"owner":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-23 17:23:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"recommended":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"all":[{"value":1,"date":"2019-09-21 12:17:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":2,"date":"2019-09-21 12:17:47.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-09-23 14:13:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2019-09-21 12:17:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":1,"date":"2019-09-21 12:17:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":1,"date":"2019-09-21 12:17:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-21 12:17:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2019-09-23 14:13:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-12 11:54:20.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-09-14 01:08:58.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-09-14 09:17:29.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-09-14 11:53:49.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-09-14 15:32:48.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-09-19 05:52:36.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2019-09-19 14:51:53.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2019-09-23 14:13:26.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-09-23 17:23:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"65556701cf9bfe0717734a0ce55f09ef11f0596e","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-11 12:34:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7442427d54505bb6c10e61a48b46e90b6177efc","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-11 12:35:34.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f2a4aa06d26813838370977cdfedae5187e145bd","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-11 12:46:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ad4eb61d9a8d843147ea04100eb76b18fbaa6283","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-11 15:45:26.000000000","message":"Patch Set 3:\n\nThis isn\u0027t going to work and I can\u0027t get a sense of what\u0027s wrong other than \"it doesn\u0027t work\".","accounts_in_message":[],"_revision_number":3},{"id":"a9959e6c398fd6d2e901f19a17b82fbb2f3135fd","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-11 16:09:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ba0574b078f6d81e238d87f1067f835fdf9b63a8","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-11 16:09:49.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"8a88fec99be2db5463e8bd8ed8211f2e4fdf5a4c","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-11 16:10:34.000000000","message":"Patch Set 4:\n\nI did some cargo culting and figured it out. There are 144 pages of docs created, so who knows if they are fully correct, but it\u0027s a start.\n\nAs people use them and find issues hopefully they will submit fixes or at least bugs.","accounts_in_message":[],"_revision_number":4},{"id":"a410e65db0c45616cdbbbd9369eeb2382f753cf5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 23:00:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1fa28e227cec423fa712068e758fbe11 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f333b8081bb4b9bb95442df5ec0e0e7 : SUCCESS in 8m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98be34988a724120b6068efdb03a08df : SUCCESS in 7m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad28d32409f44c209d22362df8b9b90a : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/506364e917bb4056a5a672886d784739 : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/43e2cca79bf64d0682c2a417179912e2 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f252cd24c924ef09cdadc3473e1f844 : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0326c07c816f4047acae44afd569ee39 : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/706470f373524feab9520d3ad78347bf : SUCCESS in 5m 39s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/88baacf4bf4c4414855a98480d8836ab : SUCCESS in 4m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8724f965ce92409dbf6952fae505a134 : SUCCESS in 4m 26s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/78384ed207384fd0aeb106b6b9f3b100 : SUCCESS in 5m 17s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/c85f2d1d8c524454b4b3d060693b81a8 : SUCCESS in 10m 44s","accounts_in_message":[],"_revision_number":5},{"id":"baefc877c70b07049a91926698451d134014e3d8","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-12 11:54:20.000000000","message":"Patch Set 5: Code-Review+2\n\nThis lgtm [1]. I didn\u0027t read every word, but the formatting and organization looks legit, and I tested a link or two as well.\n\n[1] https://storage.gra1.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_3f3/681458/5/check/openstack-tox-docs/3f333b8/pdf/doc-placement.pdf","accounts_in_message":[],"_revision_number":5},{"id":"619fb50c8df18990a7a3090dbd0736f2502f011e","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-13 10:32:16.000000000","message":"Topic changed from build-bdf-docs to build-pdf-docs","accounts_in_message":[],"_revision_number":5},{"id":"f9a0dc666682da9abd0ff0faed349e6cc84ad10e","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-09-14 01:08:58.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"df37774f78352be181d0c8f14f3c8b9747419aa7","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-09-14 09:17:29.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"b337574317fb8bdfaadfd0c557e4bcff5ad596fb","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-09-14 11:53:49.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"f05ac5d7ce67887da36978f80292a4be15524f0b","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-09-14 15:32:48.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"0a6b4551c465ea3e4c3021775f9bc10dfed35f11","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-09-19 05:52:36.000000000","message":"Patch Set 5: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":5},{"id":"ea83b538aa89af3b7f68ea586960e9f7b7eabab2","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-09-19 14:43:15.000000000","message":"Patch Set 5:\n\nI\u0027m not looking into this closely, but the section structure seems weird to me.\nThe CLI utilities, Config Guide, Installation Guide and so on are under the \"REST API\" section...","accounts_in_message":[],"_revision_number":5},{"id":"a917ab1b3e562acc16159c34bb10d2bbb220b5ff","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-09-19 14:51:53.000000000","message":"Patch Set 5:\n\nLooks like the paths in [1] under\n\n.. toctree::\n   :hidden:\n\nis not hidden?\n\n[1] https://github.com/openstack/placement/blob/885012cb646d2ff884ea903b7c8dda7cbb465c61/doc/source/index.rst#rest-api","accounts_in_message":[],"_revision_number":5},{"id":"b02436a365017cb67109c01f54487cf63ebcee1f","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-09-21 12:17:47.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"884adf1bff980d15a6493be4770a91c594770563","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 12:36:44.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b74dee16eee940f29e96f4a18936e92c : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06296df6a0134f048efa275c29adae22 : SUCCESS in 5m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e695104f956407686e7effadf083d43 : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e893f653ae346d2af2a55d4bf83b241 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dcca0f6468d4864932608e3f602c636 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/495ed5e796fe4162a1927f276e9edf95 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37120a9c30934a72a1592097c99691e7 : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b67a25eefa15436fa2a9d932f38cd283 : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49212c61306744fcb1f98347cd7abba2 : SUCCESS in 4m 32s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e9e413a543564af6be127ac4094aeb4f : SUCCESS in 4m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/41ef77694cd7431abf21ebe8a167138f : SUCCESS in 4m 22s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/10c73e8706e84d639e96bd9ab3e0b637 : SUCCESS in 4m 55s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/b30509c5bfc347f2a53ef1203982e4bb : SUCCESS in 12m 25s","accounts_in_message":[],"_revision_number":6},{"id":"eb4d3543d78edd0323a2b89344fa8cf8cf369091","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-09-23 14:13:10.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"22dd2a945cf0e61f359f238451dac0607899ed98","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-09-23 14:13:26.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"d3766d5b5ba0a57cc957b5a0bba891bfc68152e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 14:13:44.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"170ceac9ab4528aa93133244c4157abbfd663975","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 17:23:51.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8d4408b7e09a4c6f848e66e75094ddde : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94c8dff9c5b940138f134776713d342f : SUCCESS in 6m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15260ad10f564cc1a93a300699176fb2 : SUCCESS in 7m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cc79ab59c6f45c0a156f42cc5fefe72 : SUCCESS in 5m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b0ac073f07bc4173813acf81233b7250 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e10835b5e5464275bb5becda76da82e2 : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/573c5fc3d76a4346a3b4ba52dd248e09 : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c5696f620f2414d952a76cbe63e2fb5 : SUCCESS in 4m 32s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6042ed9e22fb418ea7d360f7c8b4abb6 : SUCCESS in 5m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c3240400fd7e465f821374add4894d4b : SUCCESS in 4m 48s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/0cd817972618493d8bac8c96101fc211 : SUCCESS in 5m 54s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/6300a8b10f88407c8cbf0d722d7f31f0 : SUCCESS in 24m 56s","accounts_in_message":[],"_revision_number":6},{"id":"64122342934826c7df5e419c9fd223c8cf11f594","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 17:23:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"59840ad60cc9df7eb9715f877c63b18d9c394c91","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 17:26:55.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/63678f8f95a747d7bc90a842af8a6437 : SUCCESS in 58s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/190daa67a9da40bcb919277b52b340dc : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bece9372a7af4644b1cbb632e3bcbef9 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"b93200d7d688b6d59c5cdab59eecb7f105abac43","revisions":{"fc90aa1367614feba60c9d4afc496244bf85be26":{"kind":"REWORK","_number":1,"created":"2019-09-11 12:34:28.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/58/681458/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/58/681458/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/58/681458/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/58/681458/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/58/681458/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/58/681458/1"}}},"commit":{"parents":[{"commit":"b49c67c77a4d4a379f015dbc39bcb8a893fc3429","subject":"Merge \"Update setup.cfg to include project_urls\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b49c67c77a4d4a379f015dbc39bcb8a893fc3429"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-11 12:32:59.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-11 12:32:59.000000000","tz":60},"subject":"WIP: Build pdf docs","message":"WIP: Build pdf docs\n\nThis follows the instructions [1] in an attempt to build pdf docs.\n\n[1] https://etherpad.openstack.org/p/train-pdf-support-goal\n\nChange-Id: Icf7c22bf9d1de6fb2a74a756c370930d4c00b0b9\nStory: 2006110\nTask: 35397\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/fc90aa1367614feba60c9d4afc496244bf85be26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/fc90aa1367614feba60c9d4afc496244bf85be26"}]},"branch":"refs/heads/master"},"b3f59536604a2e3826c93cab98d24c251b64a86f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-11 12:35:34.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/58/681458/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/58/681458/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/58/681458/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/58/681458/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/58/681458/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/58/681458/2"}}},"commit":{"parents":[{"commit":"b49c67c77a4d4a379f015dbc39bcb8a893fc3429","subject":"Merge \"Update setup.cfg to include project_urls\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b49c67c77a4d4a379f015dbc39bcb8a893fc3429"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-11 12:32:59.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-11 12:35:19.000000000","tz":60},"subject":"WIP: Build pdf docs","message":"WIP: Build pdf docs\n\nThis follows the instructions [1] in an attempt to build pdf docs.\n\n[1] https://etherpad.openstack.org/p/train-pdf-support-goal\n\nChange-Id: Icf7c22bf9d1de6fb2a74a756c370930d4c00b0b9\nStory: 2006110\nTask: 35398\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b3f59536604a2e3826c93cab98d24c251b64a86f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b3f59536604a2e3826c93cab98d24c251b64a86f"}]},"branch":"refs/heads/master"},"6ec9770e1a4f88b0309fe80508d16a6ff2af7523":{"kind":"REWORK","_number":3,"created":"2019-09-11 12:46:05.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/58/681458/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/58/681458/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/58/681458/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/58/681458/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/58/681458/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/58/681458/3"}}},"commit":{"parents":[{"commit":"b49c67c77a4d4a379f015dbc39bcb8a893fc3429","subject":"Merge \"Update setup.cfg to include project_urls\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b49c67c77a4d4a379f015dbc39bcb8a893fc3429"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-11 12:32:59.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-11 12:45:54.000000000","tz":60},"subject":"WIP: Build pdf docs","message":"WIP: Build pdf docs\n\nThis follows the instructions [1] in an attempt to build pdf docs.\n\n[1] https://etherpad.openstack.org/p/train-pdf-support-goal\n\nChange-Id: Icf7c22bf9d1de6fb2a74a756c370930d4c00b0b9\nStory: 2006110\nTask: 35398\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/6ec9770e1a4f88b0309fe80508d16a6ff2af7523"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/6ec9770e1a4f88b0309fe80508d16a6ff2af7523"}]},"branch":"refs/heads/master"},"46d109e8ef598a230e88312954f6b9846b3f1a53":{"kind":"REWORK","_number":4,"created":"2019-09-11 16:09:04.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/58/681458/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/58/681458/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/58/681458/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/58/681458/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/58/681458/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/58/681458/4"}}},"commit":{"parents":[{"commit":"b49c67c77a4d4a379f015dbc39bcb8a893fc3429","subject":"Merge \"Update setup.cfg to include project_urls\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b49c67c77a4d4a379f015dbc39bcb8a893fc3429"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-11 12:32:59.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-11 16:08:05.000000000","tz":60},"subject":"WIP: Build pdf docs","message":"WIP: Build pdf docs\n\nThis follows the instructions [1] in an attempt to build pdf docs.\n\nSeveral doc/source/conf.py changes are required to get this to work.\nThe most important one is\n\n    \u0027maxlistdepth\u0027: \u002710\u0027,\n\nwhich prevents the build process from stalling out and dropping the\ncaller into an interactive session.\n\n[1] https://etherpad.openstack.org/p/train-pdf-support-goal\n\nChange-Id: Icf7c22bf9d1de6fb2a74a756c370930d4c00b0b9\nStory: 2006110\nTask: 35398\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/46d109e8ef598a230e88312954f6b9846b3f1a53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/46d109e8ef598a230e88312954f6b9846b3f1a53"}]},"branch":"refs/heads/master"},"e9bd476a562762963bac8861a3dde7b9db536536":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-09-11 16:09:49.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/58/681458/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/58/681458/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/58/681458/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/58/681458/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/58/681458/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/58/681458/5"}}},"commit":{"parents":[{"commit":"b49c67c77a4d4a379f015dbc39bcb8a893fc3429","subject":"Merge \"Update setup.cfg to include project_urls\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b49c67c77a4d4a379f015dbc39bcb8a893fc3429"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-11 12:32:59.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-11 16:09:39.000000000","tz":60},"subject":"Build pdf docs","message":"Build pdf docs\n\nThis follows the instructions [1] in an attempt to build pdf docs.\n\nSeveral doc/source/conf.py changes are required to get this to work.\nThe most important one is\n\n    \u0027maxlistdepth\u0027: \u002710\u0027,\n\nwhich prevents the build process from stalling out and dropping the\ncaller into an interactive session.\n\n[1] https://etherpad.openstack.org/p/train-pdf-support-goal\n\nChange-Id: Icf7c22bf9d1de6fb2a74a756c370930d4c00b0b9\nStory: 2006110\nTask: 35398\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e9bd476a562762963bac8861a3dde7b9db536536"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e9bd476a562762963bac8861a3dde7b9db536536"}]},"branch":"refs/heads/master"},"b93200d7d688b6d59c5cdab59eecb7f105abac43":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-09-21 12:17:47.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/58/681458/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/58/681458/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/58/681458/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/58/681458/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/58/681458/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/58/681458/6"}}},"commit":{"parents":[{"commit":"76a5c83c2fd747f57e0e93aadc7d4bb0265afa99","subject":"Update nested-magic spec for root_member_of","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/76a5c83c2fd747f57e0e93aadc7d4bb0265afa99"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-11 12:32:59.000000000","tz":60},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-09-20 13:50:56.000000000","tz":0},"subject":"Build pdf docs","message":"Build pdf docs\n\nThis follows the instructions [1] in an attempt to build pdf docs.\n\nSeveral doc/source/conf.py changes are required to get this to work.\nThe most important one is\n\n    \u0027maxlistdepth\u0027: \u002710\u0027,\n\nwhich prevents the build process from stalling out and dropping the\ncaller into an interactive session.\n\n[1] https://etherpad.openstack.org/p/train-pdf-support-goal\n\nChange-Id: Icf7c22bf9d1de6fb2a74a756c370930d4c00b0b9\nStory: 2006110\nTask: 35398\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b93200d7d688b6d59c5cdab59eecb7f105abac43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b93200d7d688b6d59c5cdab59eecb7f105abac43"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
