)]}'
{"id":"openstack%2Fplacement~687279","triplet_id":"openstack%2Fplacement~stable%2Ftrain~I3df5686d13c1ce0cc38402a1317acb661ad74cec","project":"openstack/placement","branch":"stable/train","hashtags":[],"change_id":"I3df5686d13c1ce0cc38402a1317acb661ad74cec","subject":"Add --skip-locks flag to mysql-migrate-db.sh","status":"MERGED","created":"2019-10-08 12:04:12.000000000","updated":"2019-10-08 15:08:10.000000000","submitted":"2019-10-08 15:06:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687279-1570547178937-70f2cce9","meta_rev_id":"12b831e2e466d563373a33d47912ef1e7b7affdd","_number":687279,"virtual_id_number":687279,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-08 15:06:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":2,"date":"2019-10-08 12:38:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-10-08 13:40:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2019-10-08 13:40:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-08 12:38:10.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-10-08 13:40:53.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-10-08 15:06:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fa5ee2a18d52a26921d08076eaa5c826d9a040b1","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-10-08 12:04:12.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"52a71ceecb00775ac06292fbbbe7b0af62c6f520","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-08 12:38:10.000000000","message":"Patch Set 1: Code-Review+2\n\nclean and worthy","accounts_in_message":[],"_revision_number":1},{"id":"6f745c2651a0bee1d134d95d7cec49428cd5cfd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 13:22:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9ec436f4d6764b4bafc539a29d87bf2f : SUCCESS in 48m 27s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/27c721bb42094df9991c6116415fb7a6 : SUCCESS in 1h 11m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a02240eff904d1fa96941fee7178237 : SUCCESS in 6m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb2c65f3b8924b2da6a47e595febbe8e : SUCCESS in 4m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f49b2d8904b44989c5ff3635ab8863c : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/982d78d1d4524428ab58608fa171a2ce : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8ca0308b8fd6425c8bc4209a669826a4 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/363bda905e4243dfb43c44e70c59f17e : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/986c8143255d46b7ba4e7ab32ce1c9ad : SUCCESS in 5m 10s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3f596187a2f044458dd19797a0c520c4 : SUCCESS in 4m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/106dd100be664770b4ea85740bd75609 : SUCCESS in 3m 52s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/23eb91f5123b4393a578d7f2b1d0a028 : SUCCESS in 4m 51s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/598f5aa84b10414188a09a85c8f6a1c2 : SUCCESS in 11m 26s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/212ebe666a0c4bf09ecb97dfd627e764 : SUCCESS in 31m 11s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/885792d32d234c1587a9c8ae2b3293ef : SUCCESS in 11m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b8c298b05984aff8f7005f3fd91ee64 : SUCCESS in 1h 04m 30s","accounts_in_message":[],"_revision_number":1},{"id":"a3235258144cb7a50d03a2e2a1eff47c4c4f86cb","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-10-08 13:40:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nyeah, let\u0027s have this.","accounts_in_message":[],"_revision_number":1},{"id":"023a5ac46ddfde5aaa3394266d9cd444cff5c6bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 13:41:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1fd69c230261cf392b9f9483a25cdaf2764bdc5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 15:06:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/db58a501a65741e6bef130509148b676 : SUCCESS in 1h 02m 38s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/8b1acbb59e3048bbb6db57e9e5b60f3e : SUCCESS in 1h 19m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86467fb54377457cb3b1ca115ea8fc27 : SUCCESS in 7m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76cc597e6f654832983cffd9fe70b843 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f23cd7ea35f349da94a522d4a5d1ab55 : SUCCESS in 5m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/41cebcbab4df4c21a744668298bfaab3 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61649c1162bc45c6ad47028646f36a18 : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6d70fbe93e674d2bace9485512a440df : SUCCESS in 6m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/99fbd4bdfa344fe1a4f7670c50753ed1 : SUCCESS in 5m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e6421d3c633a4fc78d580c7c80a53773 : SUCCESS in 6m 47s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/cb51056977564bf1947eb015652159da : SUCCESS in 7m 04s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/8997095f519243098c1b6e8d73137b76 : SUCCESS in 15m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/daca9e55502042719b379242e9b9ba76 : SUCCESS in 1h 22m 50s","accounts_in_message":[],"_revision_number":1},{"id":"89853bdf95adbae45ce633d0d288065130cae723","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 15:06:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"12b831e2e466d563373a33d47912ef1e7b7affdd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 15:08:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51a6639861f24286b67323dde36169b0 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ba1e949595e5cf24bbdf886d4f2576487f9098f0","revisions":{"ba1e949595e5cf24bbdf886d4f2576487f9098f0":{"kind":"REWORK","_number":1,"created":"2019-10-08 12:04:12.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/79/687279/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/79/687279/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/79/687279/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/79/687279/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/79/687279/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/79/687279/1"}}},"commit":{"parents":[{"commit":"3f1a89c7dd34c53d5bc9f1190e48ac3a9320bc70","subject":"Fix domain name in install doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/3f1a89c7dd34c53d5bc9f1190e48ac3a9320bc70"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-10-07 13:49:47.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-10-08 12:04:12.000000000","tz":0},"subject":"Add --skip-locks flag to mysql-migrate-db.sh","message":"Add --skip-locks flag to mysql-migrate-db.sh\n\nThe --skip-locks flag can be used along with --migrate in\ndeployments where table locking operations can\u0027t be performed.\nFor example, Percona XtraDB Cluster only has experimental support\nfor explicit table locking operations and attempts to use locking\nwill result in errors when PXC Strict Mode is set to ENFORCING.\n\nChange-Id: I3df5686d13c1ce0cc38402a1317acb661ad74cec\nStory: #2006670\n(cherry picked from commit 47e31b894e12ca7cc8c476e5b3a194d7d0e15884)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ba1e949595e5cf24bbdf886d4f2576487f9098f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ba1e949595e5cf24bbdf886d4f2576487f9098f0"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
