)]}'
{"id":"openstack%2Fplacement~704799","triplet_id":"openstack%2Fplacement~master~Ia0aa7fc7b7f8e3a14c3e8c3292f042d7ace86bb5","project":"openstack/placement","branch":"master","topic":"bug_2007232","hashtags":[],"change_id":"Ia0aa7fc7b7f8e3a14c3e8c3292f042d7ace86bb5","subject":"Add prefix \u0027/placement\u0027 for listing resource providers","status":"ABANDONED","created":"2020-01-29 13:11:48.000000000","updated":"2020-01-30 14:46:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cfce9fa822a41287794ea86621e8a26b9ea31120","_number":704799,"virtual_id_number":704799,"owner":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-01-29 13:22:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"date":"2020-01-30 03:08:11.000000000","_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"all":[{"value":-1,"date":"2020-01-29 16:01:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-01-29 13:15:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-29 13:15:48.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2020-01-29 13:22:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-29 16:01:24.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"}],"messages":[{"id":"fecd683c4904affd9b85bf0c3979827ca16d1814","author":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"date":"2020-01-29 13:11:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2e0b84bd798d6d12f0e91d5f9e47276d1d81693","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2020-01-29 13:15:48.000000000","message":"Patch Set 1: Code-Review-1\n\nsee my comment on the story, I don\u0027t think this is a good idea","accounts_in_message":[],"_revision_number":1},{"id":"8abd5735839f33e4052cc9d6b9fb1c7d5eb416f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 13:22:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/890f2afd0d5249fcac0eb4b114727354 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83b755bf603f43a59f2d933bfd37c627 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac94f3e43d9b438eb0f7779af0fa3936 : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":1},{"id":"952e8f4f599adb7fd28ccd8c476063c5fedb655a","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-01-29 16:01:24.000000000","message":"Patch Set 1: Code-Review-1\n\nAgree with Chris. This has been brought up in the past.","accounts_in_message":[],"_revision_number":1},{"id":"bf61c855b5c31eaf8c82da9fc3e4bc47ee13d297","author":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"date":"2020-01-30 03:08:11.000000000","message":"Patch Set 1:\n\n@Chris and @Eric, links in other examples (like `show-resource-class`) have prefix \u0027/placement\u0027. Should we unify these?","accounts_in_message":[],"_revision_number":1},{"id":"964eebc9bc5da6ae8b2089273a7e393f9fa9d770","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2020-01-30 10:03:11.000000000","message":"Patch Set 1:\n\nI don\u0027t have a strong preference: the point I was trying to make in my earlier comment was that the link URLs could be pretty much anything, so the examples could have anything in them.\n\nIf people are in favor of consistency then go for consistency, but I don\u0027t think it particularly matters for the samples. In fact, in some sense, the lack of consistency demonstrates the point/value of the links: they show the way, whatever that way might be.","accounts_in_message":[],"_revision_number":1},{"id":"25ca02debfd5192955569efee9cd3657395085fe","author":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"date":"2020-01-30 14:46:00.000000000","message":"Abandoned\n\n@Chris, i agree with you, so abandon this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e52bc789b4be9de529d4d3661e5e87b25526f5e3","revisions":{"e52bc789b4be9de529d4d3661e5e87b25526f5e3":{"kind":"REWORK","_number":1,"created":"2020-01-29 13:11:48.000000000","uploader":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"ref":"refs/changes/99/704799/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/99/704799/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/99/704799/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/99/704799/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/99/704799/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/99/704799/1"}}},"commit":{"parents":[{"commit":"971c7aa76e1a3455acf8cf2210b3b7af29846696","subject":"Provide more accurate links in doc/source/user/provider-tree.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/971c7aa76e1a3455acf8cf2210b3b7af29846696"}]}],"author":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2020-01-29 13:10:20.000000000","tz":480},"committer":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2020-01-29 13:10:20.000000000","tz":480},"subject":"Add prefix \u0027/placement\u0027 for listing resource providers","message":"Add prefix \u0027/placement\u0027 for listing resource providers\n\nChange-Id: Ia0aa7fc7b7f8e3a14c3e8c3292f042d7ace86bb5\nStory: #2007232\nTask: #38508\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e52bc789b4be9de529d4d3661e5e87b25526f5e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e52bc789b4be9de529d4d3661e5e87b25526f5e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
