)]}'
{"id":"openstack%2Fplacement~722961","triplet_id":"openstack%2Fplacement~stable%2Fussuri~I2418133c7ea010a309b759a87b03d9dd28a007b0","project":"openstack/placement","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"I2418133c7ea010a309b759a87b03d9dd28a007b0","subject":"Update .gitreview for stable/ussuri","status":"MERGED","created":"2020-04-24 18:00:05.000000000","updated":"2020-04-30 20:45:59.000000000","submitted":"2020-04-30 20:43:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722961-1588279428505-ae45bbcb","meta_rev_id":"84ef84690b228419071a241f03892b90a3ddfd4c","_number":722961,"virtual_id_number":722961,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-30 20:43:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":2,"date":"2020-04-30 17:02:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":1,"date":"2020-04-30 16:24:51.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-04-30 17:25:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2020-04-30 17:25:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-30 16:24:51.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-04-30 17:02:28.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2020-04-30 17:25:54.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2020-04-30 20:43:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"65fc8c26609e818d8221c82557f07f49f7f90a96","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-24 18:00:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5794aa18e9f159b3da1632c94e3ac306af6a10ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 18:59:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1b5331b2a204956a99c1e45062e0376 : SUCCESS in 9m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b88df65c6df946d4bc643220ef019992 : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf0df106e7e64a20ab396351fcaf3654 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90ed8d6aaec040c9bf70e0b17011ff6b : SUCCESS in 8m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c32a76db21e4251b4831ae0ecac727d : SUCCESS in 9m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1eeabe76b9b24aaca0eb904c5d02c084 : SUCCESS in 11m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d0f73a5adf7d428c9d59e2368f09eb50 : SUCCESS in 7m 59s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/0dd71543d8af4bf7aed3cc784b4d9b6c : SUCCESS in 7m 11s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/812b3d18c8904e92901c752518bd565b : SUCCESS in 16m 29s","accounts_in_message":[],"_revision_number":1},{"id":"16e9e70028d4fead937d49efcd1621dc84449407","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-04-30 16:24:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"006c0ffbc0ce7882b23e85a1099b6e46cbb97dc0","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-04-30 17:02:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a909abe48a85374faae1d6bdb3f83b5f72dc2e3e","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2020-04-30 17:25:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"12471106e80e3da1a9541e60a1b1245d727edc77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 17:26:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"92aff6fd3bfa6e8b22708c9f1906692c8262e048","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 20:43:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"61bff58a3db238453b3d6b389fa904b9adcaa398","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 20:43:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/18528712be72435699a294850d90d0d2 : SUCCESS in 1h 14m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ea57dba396940d7993fce644e065a6e : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6edc2872319c44b8a7a6d3307ebd7dcc : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff8c80147a3b46b98265420612e56499 : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/053789af2882498a9cfb729d934b83fa : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69b9e48869b74842b53e38070a0057e7 : SUCCESS in 6m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0fb9c9d536fb43c3a3e40ee53c4bd919 : SUCCESS in 5m 46s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/b1d7673976454ff0a7db88b559e02deb : SUCCESS in 5m 46s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/cbd693ef4b7149b4928618a223afe853 : SUCCESS in 15m 46s","accounts_in_message":[],"_revision_number":1},{"id":"00032a7836262304fb5eafcec957a74d95a7a1b3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 20:45:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93b450ec22b446a1b8de1facbf2d8f94 : SUCCESS in 1m 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a8f981e36ac5e975ceb15f762ecdd34161bc4139","revisions":{"a8f981e36ac5e975ceb15f762ecdd34161bc4139":{"kind":"REWORK","_number":1,"created":"2020-04-24 18:00:05.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/61/722961/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/61/722961/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/61/722961/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/61/722961/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/61/722961/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/61/722961/1"}}},"commit":{"parents":[{"commit":"0f90d197e11d7aa605c0e1d338db1d146a00d8eb","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/0f90d197e11d7aa605c0e1d338db1d146a00d8eb"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-24 18:00:04.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-24 18:00:04.000000000","tz":0},"subject":"Update .gitreview for stable/ussuri","message":"Update .gitreview for stable/ussuri\n\nChange-Id: I2418133c7ea010a309b759a87b03d9dd28a007b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/a8f981e36ac5e975ceb15f762ecdd34161bc4139"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/a8f981e36ac5e975ceb15f762ecdd34161bc4139"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
