)]}'
{"id":"openstack%2Fplacement~740576","triplet_id":"openstack%2Fplacement~master~I9ef2b3bb7f8cea35461cc4569430ff48ee1272c0","project":"openstack/placement","branch":"master","hashtags":[],"change_id":"I9ef2b3bb7f8cea35461cc4569430ff48ee1272c0","subject":"Replace assertItemsEqual with assertCountEqual","status":"MERGED","created":"2020-07-12 03:39:19.000000000","updated":"2020-09-09 14:05:54.000000000","submitted":"2020-09-09 14:00:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740576-1599660011576-7f25b096","meta_rev_id":"fa7639032e93b2dc571b0099be28caaf9475b86f","_number":740576,"virtual_id_number":740576,"owner":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-09 14:00:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"recommended":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":1,"date":"2020-08-22 03:24:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":2,"date":"2020-08-06 15:20:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-09-09 12:46:11.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2020-09-09 12:46:11.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2020-09-09 01:19:36.000000000","_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"date":"2020-09-09 01:19:36.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"date":"2020-09-09 01:19:36.000000000","_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"date":"2020-09-09 01:19:36.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"date":"2020-09-09 01:19:36.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-09 01:19:36.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-06 15:20:15.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-08-22 03:24:50.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2020-09-09 01:19:36.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2020-09-09 01:19:36.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2020-09-09 01:19:36.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2020-09-09 01:19:36.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-09-09 01:19:36.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2020-09-09 01:19:36.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2020-09-09 12:46:11.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2020-09-09 14:00:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8204d38e9281d49362be1866df0584b64f6a08d2","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-07-12 03:39:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc9359d80b54148db738e0ab94c568008f743f30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-12 03:48:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bf9d22f48d54a3ba6cd8ae2b42cbc1b : FAILURE in 8m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50ad1efbde07472aa83ef1798554e406 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/485af393bef44b14b9f9b5e029c01103 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/acccd6b97d8749f482c4ae91b01ad09c : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50879267ade347c58900bf59bd352029 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/196291e7ea8d46ddb987166a80a684f5 : SUCCESS in 8m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/11a55b5d4c75462fb559db2031b9023f : FAILURE in 4m 28s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/b51bf05a6f1a4ec2950234677df77362 : FAILURE in 5m 16s","accounts_in_message":[],"_revision_number":1},{"id":"76724174593c793144136f955b1092517410ea95","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-08-06 15:20:02.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d79f1f016602cc6d10b18b1af51e79959d221247","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-08-06 15:20:15.000000000","message":"Patch Set 2: Code-Review+2\n\nI don\u0027t think the failures are related","accounts_in_message":[],"_revision_number":2},{"id":"29fcf3ee201afea32f0f82ad75f09eec7f53975e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 15:34:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca60354bf7464e6bac790e81bedf98d9 : SUCCESS in 7m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8d0a0a91e0a4f51bb37c5a98fea1d39 : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1128d421ee1144e9ae64befbde35a8df : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/969f988efb9f4c36ada1693b420ff6e7 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f83423d36ca044d599ba1b6724e1baf2 : SUCCESS in 7m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9a7d37b78cd4dcfb3eb1a6b3e265b2f : SUCCESS in 9m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d1df761209eb4c70904a6190f00c63a9 : SUCCESS in 10m 42s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/19293e18818f4157bc97e0a1c41f4e44 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":2},{"id":"5f802323988a69bff66e0df14d8a289fbbe88e65","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-08-22 03:24:50.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2274408a6af519cdc83e286c5e0bc3c5ac2c7657","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-09-09 12:46:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLe ugh.\n\nThis patch is good. But what a terrible name choice by the unittest maintainers. ``assertItemsEqual`` clearly indicates that it will assert that the items are equal. ``assertCountEqual`` *sounds* like it\u0027s going to validate the *size*. I had to go read the docs -- and will likely have to do so every single time I come across this method -- to validate that it\u0027s still diffing the *items* (regardless of order).","accounts_in_message":[],"_revision_number":2},{"id":"3eb118272944a0d7f762c448e6e7906eeeececb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 12:46:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d9f9fb4c6b4b5bbcdb708be1a9425ab2080e2416","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 14:00:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe5d5d539691442f971153721a69b418 : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59b1ab2e9456480d8f5044b4b77fd326 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e93ddbbeb2644fa1adc5e72dc37e366b : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/841a035257304ae68bd9361e22e230d6 : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c71adb1a2b9459fb1bfdee404e78077 : SUCCESS in 9m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d51ec21c154d4a3dbc9c2921d58ae89d : SUCCESS in 6m 05s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/5076e917b26b4cef88ff53bbb59437d7 : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":2},{"id":"1ba32662e586561f9bf2e892956041db1994c825","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 14:00:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fa7639032e93b2dc571b0099be28caaf9475b86f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 14:05:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b69eb3229ff46f693945e3b38f9c951 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6774a7507f5504135fd03de6befe4f3ec6ee96d7","revisions":{"65ba196a39a863d6f91328f5b9683276bdd3d97b":{"kind":"REWORK","_number":1,"created":"2020-07-12 03:39:19.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/76/740576/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/76/740576/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/76/740576/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/76/740576/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/76/740576/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/76/740576/1"}}},"commit":{"parents":[{"commit":"df56e21c906ecc5b145df015f740dad885511538","subject":"Merge \"[Community goal] Update contributor documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/df56e21c906ecc5b145df015f740dad885511538"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-07-12 03:38:53.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-07-12 03:38:53.000000000","tz":480},"subject":"Replace assertItemsEqual with assertCountEqual","message":"Replace assertItemsEqual with assertCountEqual\n\nassertItemsEqual was removed from Python\u0027s unittest.TestCase in\nPython 3.3 [1][2]. We have been able to use them since then, because\ntesttools required unittest2, which still included it. With testtools\nremoving Python 2.7 support [3][4], we will lose support for\nassertItemsEqual, so we should switch to use assertCountEqual.\n\n[1] - https://bugs.python.org/issue17866\n[2] - https://hg.python.org/cpython/rev/d9921cb6e3cd\n[3] - testing-cabal/testtools#286\n[4] - testing-cabal/testtools#277\n\nChange-Id: I9ef2b3bb7f8cea35461cc4569430ff48ee1272c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/65ba196a39a863d6f91328f5b9683276bdd3d97b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/65ba196a39a863d6f91328f5b9683276bdd3d97b"}]},"branch":"refs/heads/master"},"6774a7507f5504135fd03de6befe4f3ec6ee96d7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-06 15:20:02.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/76/740576/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/76/740576/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/76/740576/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/76/740576/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/76/740576/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/76/740576/2"}}},"commit":{"parents":[{"commit":"4b95c078cc75987979a96bceafce152d6153adcf","subject":"Update perfload jobs for python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/4b95c078cc75987979a96bceafce152d6153adcf"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-07-12 03:38:53.000000000","tz":480},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-08-06 15:20:02.000000000","tz":0},"subject":"Replace assertItemsEqual with assertCountEqual","message":"Replace assertItemsEqual with assertCountEqual\n\nassertItemsEqual was removed from Python\u0027s unittest.TestCase in\nPython 3.3 [1][2]. We have been able to use them since then, because\ntesttools required unittest2, which still included it. With testtools\nremoving Python 2.7 support [3][4], we will lose support for\nassertItemsEqual, so we should switch to use assertCountEqual.\n\n[1] - https://bugs.python.org/issue17866\n[2] - https://hg.python.org/cpython/rev/d9921cb6e3cd\n[3] - testing-cabal/testtools#286\n[4] - testing-cabal/testtools#277\n\nChange-Id: I9ef2b3bb7f8cea35461cc4569430ff48ee1272c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/6774a7507f5504135fd03de6befe4f3ec6ee96d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/6774a7507f5504135fd03de6befe4f3ec6ee96d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
