)]}'
{"id":"openstack%2Fplacement~766785","triplet_id":"openstack%2Fplacement~master~I942da4b1880cf2312af2f8e15aef9f33be991f87","project":"openstack/placement","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I942da4b1880cf2312af2f8e15aef9f33be991f87","subject":"Pass context objects to oslo.policy directly","status":"MERGED","created":"2020-12-11 20:06:52.000000000","updated":"2021-02-02 21:40:33.000000000","submitted":"2021-02-02 21:38:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"766785-secure-rbac","meta_rev_id":"8af34bd0217b7816731811fd7d0697c7e665f6d5","_number":766785,"virtual_id_number":766785,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-02 21:38:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"recommended":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2021-02-02 21:40:33.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2021-02-02 21:40:33.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-02 18:40:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-02 18:40:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-11 21:31:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-08 10:37:58.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-01-21 17:52:37.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-01-22 18:52:44.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"b6a5e4e4a9d5ee0c3f93d47d0d9851510fa24214","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-12-11 20:06:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5dc9560c6ff4252c69b95d3bc83189c0f8cda923","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 21:31:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/4020f00789854c4c9e230bc25f1b340e : SUCCESS in 1h 08m 53s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/69f2f438b93e4a13b48ac77a7c70e0cb : SUCCESS in 1h 18m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f02ba73da5dd41268ac974c1f9492ce3 : FAILURE in 5m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77e502fc2a1244519bf198dcdc0d5f86 : FAILURE in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55c311138bdd4b98adae9e5cd4ec279e : FAILURE in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3768f24ed3f24facbed815771e98efa0 : FAILURE in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05ffda2de5dc4404876f90f07e6371a4 : FAILURE in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bcfd952e5503487ea9d4d963db96a949 : FAILURE in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe7438bbf94c4887a2e3136b989c0f5e : SUCCESS in 7m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f70497247fb9483eab5bd373bc6e7aa4 : FAILURE in 4m 17s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/a7abd4b2ab134112a61f05c0b98e7a6f : FAILURE in 4m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/442a6f4399584170be5ef2d7ab596b1d : FAILURE in 4m 20s\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b21d2819c7e341ca9d212c794f62ee13 : SUCCESS in 14m 32s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/3f4b3036494e4d6fac2398c8ddedda44 : SUCCESS in 32m 38s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/837fa23827a64fa496b84e91bbeb067b : SUCCESS in 10m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad1145160b16476ab445eaf995225f51 : SUCCESS in 1h 05m 33s","accounts_in_message":[],"_revision_number":1},{"id":"b4ccf91f204fbeb4d8ab78aa255d06f24e300d19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-12-14 15:12:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0ad03712c93698ceb928fd5ba55613f029407571","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 17:47:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d12f7af2a10c4627973b2a7b4de33e41 : FAILURE in 3m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/79df24c4d1f44f61bb81dada1ab27a92 : SUCCESS in 1h 10m 33s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/2a20ab949628443db468aafd14c501b7 : SUCCESS in 1h 18m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/297cd3635bf246579249572b8496a0de : FAILURE in 4m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ad82c320b5445f0a55a7045a0db7ea1 : FAILURE in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ddb6034a6af470d8d55944e9ef4bd3f : FAILURE in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c059be39a4a4566b98d85cc60a4b9bd : FAILURE in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be4a394b7ec5477dabee8b2a6ba75b8b : FAILURE in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be31350542684a35a1a7b03b48045d7a : FAILURE in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5584dfc76b0e4aaca030a32691f41067 : SUCCESS in 8m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/980d1b2e65b54f1eb1f612192c4ab983 : FAILURE in 5m 03s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/e292399ee85a4dca8c0cf81bf4e11bb4 : FAILURE in 5m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/02208b250276439fb88909743efe7f4e : FAILURE in 4m 30s\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e38be1d551af4499b42d6ba80bcc519a : SUCCESS in 13m 55s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/14f9dbb1e44748d0a61d2838c67c2d4e : SUCCESS in 31m 30s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/7d1150db69784d4e9a0cb97f82241f76 : SUCCESS in 12m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9acc7d9be80f47afb9025bb800fdb510 : SUCCESS in 1h 06m 53s","accounts_in_message":[],"_revision_number":2},{"id":"04f90b32def82f1fa5584947b790be54ca215003","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-12-14 20:05:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0d9146260e84c10cf40a5af9c3a10d9b13b7c3d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 21:23:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8d64d0648a9240c981cb393f800f0a63 : FAILURE in 3m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/0a4f56b04bba4b73b1bd62202537585a : SUCCESS in 51m 11s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/3e1c8068015e4a19bfc935ddb8081026 : SUCCESS in 1h 15m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ba55aa7e633432994a130ecfb6d1b56 : FAILURE in 15m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/335c19ecb4f448b782a9756eca038d5c : FAILURE in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75ff9de503ee4197a93f5eb808873a56 : FAILURE in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/444cf465270e414f9a06bbb4efba41a9 : FAILURE in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9be80068370243fd9dc7ade56938257e : FAILURE in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c36c720e00c6439386f6fc56d1286b1f : FAILURE in 4m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48c9ded0b042451d82e0e951a46e978a : SUCCESS in 7m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e552ec6a8de84711b6a837d005c713fd : FAILURE in 4m 30s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/0b62f82711624b8aac81250d19dd944c : FAILURE in 4m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/719b3a3b0359466783210c195f3e1b4e : FAILURE in 5m 07s\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9c85df2d4e0c4156a9e9dfb10871e1f9 : SUCCESS in 12m 03s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/d8d362666fdc41df945f5296acc0701f : SUCCESS in 37m 30s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/47a46bb7bafa4662bfc8955045e32a38 : SUCCESS in 12m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c31eab37efc4c2c8dfc2b17b420e401 : SUCCESS in 1h 03m 34s","accounts_in_message":[],"_revision_number":3},{"id":"1c64f05696317668477b38073841c906ec5586f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-12-15 00:24:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8066421cce23e41f038d6e35f33c9efc5e885ceb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 01:39:18.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/07a0992a53f3483fbc0c4b5bdaadce9e : SUCCESS in 4m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/77516736b1ec4ea0be1e4c13923e7f1f : FAILURE in 41m 58s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/640f174882644a2ab92d970d6cc8fa1b : SUCCESS in 1h 13m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d8966c8a86c49f2a684e46afdb0a1c8 : FAILURE in 5m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a5b3587c94d4eecbe7fb130ae4980d1 : FAILURE in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec0be1155a61496fa0a8eb408d2c5d14 : FAILURE in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cba1e50df2b4aca83bb992af8960e68 : FAILURE in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9439c2d8bbe14b1eb0ee98313b0d9c75 : FAILURE in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a3ea2e0e1824aa6b7f09b97ac5da729 : FAILURE in 4m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1c3f2dc7bbc492ea5c98c519ab32b91 : SUCCESS in 7m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/63e701bf631342949cd6ef37f4e0e8a6 : FAILURE in 3m 47s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/0dd424a4497e4ed29ae9813a4376bdf3 : FAILURE in 4m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a7f23d49ba2a4c6786b84ca71bdeb456 : FAILURE in 4m 18s\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4c0a806079fd453182a130531851aff1 : SUCCESS in 12m 27s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/e1e77d436f7f4ae2b0c1bb9a04c4b3cd : SUCCESS in 17m 43s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/6d4e0c970df4446ea132e07d8991f100 : SUCCESS in 12m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9fe8174cf6a841e1bde05c7d619f1e0e : SUCCESS in 52m 19s","accounts_in_message":[],"_revision_number":4},{"id":"dd4e16a12338908d37634f3af5ae8b7da28e6366","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-12-16 13:18:47.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"7b7f9ddf5897e34ac3aa2aaa0cb302665ee2f9f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 23:53:06.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d7c6cffb17bc4a85b2612dc75c142326 : SUCCESS in 4m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/b4f3f479a8d84427a10409cda3575446 : SUCCESS in 44m 34s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/0a38e4e4ce68447e8121735b26e1df82 : SUCCESS in 1h 13m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1b3d02d605846a196b9661b850dd2c1 : SUCCESS in 5m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29b1c898c9754ab78689fbca06569e49 : FAILURE in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/574720e4e4534d639085945d8c409891 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edc96bfe08384908a731b1fcdcf462f5 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0b8621e235b4bd5a710eb7b2556447b : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f9770dc3afc4a389a72e514a537a122 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bb1443485ea4450903de461862b712f : SUCCESS in 7m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dda16f63ce3b4a66a0ebb2c062a82344 : SUCCESS in 5m 34s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/03ac54702c1748928be5f97208677c0e : SUCCESS in 6m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bcfd0938a35548cb989c5485c2e36648 : SUCCESS in 7m 10s\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c5e0fd94efaa448d86a2eeac2e78bc7c : SUCCESS in 17m 53s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/b18c6b3d2ddb48fea3441038ac3ef0c0 : SUCCESS in 34m 16s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/4de5ea6473ec40258cf1b4a86f2b0141 : SUCCESS in 12m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66385b9a0c8b4b05b507fd2273605732 : SUCCESS in 1h 06m 32s","accounts_in_message":[],"_revision_number":5},{"id":"9946b23669243534210cfe9dabaef35ee97281a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-12-17 14:53:52.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"6bde3bcfbb5d09dd74f9697ad88924707c7ac1bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 22:14:11.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6a976b3a40a748a59e2a89b535d1eebc : SUCCESS in 3m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/88c1b3db0b5e413c8e81f70599238644 : SUCCESS in 57m 14s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/b92099d1403c4185bd144656c45b535d : SUCCESS in 1h 27m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8423ab4a9a4546eaaaca14d521af5a27 : SUCCESS in 5m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b931425f9c0d49fe89a1156fe1c09bb8 : FAILURE in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/579606cf87324c2582c7ae1f36b7ea39 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b392cbd41d0f43e6a4e1fdb490a58d6b : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/956e5de124ee42dcb8550e340283ace4 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88b4323099674f86900f77b32e180859 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1daf18bd72994c7986f0ae30db064ce7 : SUCCESS in 7m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/26b3359e3a3642929cbccd6fdb6075b6 : SUCCESS in 4m 47s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/25b5d45980ac4f6e9f066ec58bd4931d : SUCCESS in 5m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/da5abb213850441bafd070ebca86e5cb : SUCCESS in 7m 49s\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f300c93453ab460b98c192dab41ded91 : SUCCESS in 16m 07s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/34338fa72f964b7f87458fbf4d51e520 : SUCCESS in 25m 55s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/91b1c59d483345cb964b972405fd434f : SUCCESS in 13m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d1bf95b4cdd476b91eb5a933694bee8 : SUCCESS in 55m 40s","accounts_in_message":[],"_revision_number":6},{"id":"af2a8b937dbb2cebb256f0343240084cb7010bcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-12-17 23:07:51.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"58b1305e1e75e66b6dddc3cc3186383464cf24d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 02:15:13.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/50631820d3544a3eaab74d6edb429ade : SUCCESS in 3m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/6bc033cbbdcd46efab68b775261c32ce : SUCCESS in 1h 02m 29s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/2f1351ba2b1c4fbf8abda897aeaf140d : SUCCESS in 1h 24m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6622b3a778f4178b3d1435925bfadb2 : SUCCESS in 5m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d0b77c3bac24b3bb579a9001102ae49 : FAILURE in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2a94234db1847c695a7a40b53f37944 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d29a1d5b0fa546b3bca5ba4763118abf : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1168045a8a14caaab57c8d1e7fd34de : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99c4a3b479e54c1a830449b3b277fca0 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6cd6fe103094aa98e07614caee68aec : SUCCESS in 7m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a7db8a9688704487ac6b978fbdf2ceba : SUCCESS in 6m 12s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/820352557e5f4a23bcf16e54f3877dbf : SUCCESS in 5m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c5e31bd8afce46358a23ff6031d5857f : SUCCESS in 5m 24s\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b525b50f501b417097db077d52e2ead2 : SUCCESS in 15m 42s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/049a8b4d8a134b708f37b5063490671e : SUCCESS in 28m 58s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/a125cf4e7a4444e591dce1b58696604f : SUCCESS in 13m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a02ca1acd4944d78a55b243646f94a7 : FAILURE in 2h 21m 14s","accounts_in_message":[],"_revision_number":7},{"id":"42a954fad91f7c15b44f7c791f6548f4f8d933c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-01-07 04:46:24.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"8b5314fd2c0d510abba0bb78ad4bcfa133b22590","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 05:51:51.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/25ce4eef2cca450daa7c4940cdc209fc : SUCCESS in 3m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/7cbc767b56dd4f4f9326e36b94fba759 : SUCCESS in 47m 14s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/9f3c8a2d167d46cd986f3c9b422393e3 : SUCCESS in 1h 02m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10b6e73c84c34b3cb3899085c17fb3ec : SUCCESS in 6m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57b6ac3003434bda91db5d32d471004e : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/046f434ec44b44728f3a9cee2b5af04f : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6afac823d1140d0860124be11379b55 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ff4e38cb227497791d6293c2172689a : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1cedf01ace0d45889809d64ba816b961 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3115d8efed14b53a31a76bb35c348f4 : SUCCESS in 7m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/36f00d1777244972b3890a8b1c30ff78 : SUCCESS in 5m 38s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/03b70bcb16ce4e36b41a55f889a62daf : SUCCESS in 5m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8a669e145e064f94b1d4fd5e360997c6 : SUCCESS in 5m 52s\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/24f8dd6da3884fb59a610fdf4b87fd96 : SUCCESS in 13m 22s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/a953df3e604c4b9888cdf8f400608e95 : SUCCESS in 17m 55s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/8a4319475559402f9c99680758fd0898 : SUCCESS in 9m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/310cd2a601714e8eac7e2bce2c98df82 : SUCCESS in 1h 01m 38s","accounts_in_message":[],"_revision_number":8},{"id":"211881e66b3b390367573f2aa747baf9f92b51bb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-01-08 10:37:58.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"2f56b655d580b6325b5f7b44ed4f8aa57d09dbda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-01-21 02:26:02.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"a24793cbd9de94f70bf350e6b8de0aa4b6f55711","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 03:51:46.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/6bd01510e45d476ab4be0b278be46502 : SUCCESS in 1h 07m 06s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/31c3405ad7de4a17984fbb922eacf4ae : SUCCESS in 1h 22m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4dfe2e5c3af6456b90df8be2e3655125 : SUCCESS in 7m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5f8b9a9f24a4e2495a24541c8f76faf : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d090e63e9dc47ee8d7e356e9c5de227 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/624b85c8cc844206be1fcaf798e3eb64 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e924bc3c5ab4f368aa3725b6fd11473 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a2c9cf621ef4e4984eb43d2068e7c6b : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eae97952c6414050a97d59815bfdef2d : SUCCESS in 7m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/662256ff53c048668b5db9ba1bae1e7c : SUCCESS in 6m 21s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/40760ef3a269400c94376bd30cdcae38 : SUCCESS in 5m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a9ad6b28718d4599a2e0099ab10ae0b3 : SUCCESS in 5m 53s\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bb77862c19f749a59bd48ec871e56593 : SUCCESS in 14m 34s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/9b28835b3c9e46d5b9f840cd4260739e : SUCCESS in 22m 21s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/8a38a76e06344ed2a0e3f4e6acf9263a : SUCCESS in 10m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d97106b7b6f14b10ac04c91e36cdf8a9 : SUCCESS in 1h 08m 55s","accounts_in_message":[],"_revision_number":9},{"id":"35644b42d5df75c4d0347c7a0d674ea1ab10d156","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-01-21 15:14:36.000000000","message":"Patch Set 9: Code-Review+2\n\nConflicts due to l-c fixes, I suspect. Change is still fine","accounts_in_message":[],"_revision_number":9},{"id":"6a5858fc625c93074342d79836eb2d7f49ee579a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-21 17:52:37.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"13a55ece08c0f2e06a0af26a0b8cde56f7c5b8ab","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-01-22 18:52:44.000000000","message":"Patch Set 9: Code-Review+1\n\nthat is a very useful feature of the enforcer","accounts_in_message":[],"_revision_number":9},{"id":"1425a9274cc3496406db0c810d438a9fdd2f7c77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-01-27 17:35:11.000000000","message":"Uploaded patch set 10: New patch set was added with same tree, parent, and commit message as Patch Set 9.","accounts_in_message":[],"_revision_number":10},{"id":"c95cb68a57d5437451e0a18264fe9368a85b5ed4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 22:12:03.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/a011eb1862a34b0dac5bc311feac65f1 : SUCCESS in 56m 58s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/265783c1014440349c0ed80407adbb7b : SUCCESS in 1h 15m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/036da9e9c7184aaba3ff946fe60d2c9f : SUCCESS in 5m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dcf50443f59042ed9cffb7bd3578a411 : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b6e46f8c60d443aaf564c7f379b67aa : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/298760344eeb4e6a98f7a96c9d9e77d0 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5c770fc8ec14c419a7f10ea89670e2f : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/602c2bcb6e9c4d958645f3e4ecd5eb5e : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3574afec2504ae4b0ee7e7a0faa9b02 : SUCCESS in 8m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6dd75f15cdc0421bb274ec2ac992a486 : SUCCESS in 5m 04s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/fcddf2ec71044723b21a42cf533df0ab : SUCCESS in 5m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b17f9b203c0d4d4898f87dd3371d1268 : SUCCESS in 5m 49s\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/697362a8b0b84c7cb36767b28c95f64b : SUCCESS in 14m 26s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/4992293a4e5844ab94b2b83fcce296f5 : SUCCESS in 24m 34s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/15254e382a12405d8ab50c945f52e5e8 : SUCCESS in 10m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad36cbe38024431991fd0e758683b561 : SUCCESS in 58m 57s","accounts_in_message":[],"_revision_number":10},{"id":"11388c3c7fdf84ef2ea2c8bf0aac6f89aec33620","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-02 18:40:33.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\nconverting my +1 to +2.","accounts_in_message":[],"_revision_number":10},{"id":"e1abd39c91a44694bbba4c9db9f3e98a03859be7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 18:41:10.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"11abe9ba5f9a1f21299f948855f14a61eecd5385","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 21:38:56.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/ed36825fb0c34b32aab13f288bd51dbf : SUCCESS in 1h 00m 35s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/e70daa4339d447f8830fb012fc8bc235 : SUCCESS in 1h 31m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3578c06c84714ca19bf64cd87254e7fa : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/965e42e777b647cbb919642be268c775 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38dbc370ef2947ba95202efd9e7611cc : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91b9c14db3f9474d865a753b7c36026e : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42a1ced1145043b08fff8c552a7d0df4 : SUCCESS in 9m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/11621287798a4491b03c6b48daea12a4 : SUCCESS in 5m 38s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/430af9edc9304235add8a675d729b32c : SUCCESS in 6m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/29c85acd704a4bb2905ca41d0f91120c : SUCCESS in 7m 00s\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e8b9aa65a9e04f299098112b864c61c1 : SUCCESS in 16m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0332bb0333c846b2b0d2aa669d446b89 : SUCCESS in 1h 01m 36s","accounts_in_message":[],"_revision_number":10},{"id":"6fc9f4183b9c47ea7d9b2caa0cf691167422d1b9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 21:38:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"a1b216ce2707a625395d14aeb8d45b928b5d90e4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 21:40:33.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1458e1f19b44e8cb7dad0f3f2ae175f : SUCCESS in 54s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"560de6a37f7bb9511be11b00c2adb2a03435360f","revisions":{"b7e364504f25554a9eb0eb7264b9a6c09408e8d6":{"kind":"REWORK","_number":1,"created":"2020-12-11 20:06:52.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/85/766785/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/85/766785/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/85/766785/1"}}},"commit":{"parents":[{"commit":"f0442937d9b91a1e0dc182a82520bab0b3cba7a1","subject":"Bump oslo.log version to 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f0442937d9b91a1e0dc182a82520bab0b3cba7a1"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-11 19:58:34.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-11 19:58:34.000000000","tz":0},"subject":"Pass context objects to oslo.policy directly","message":"Pass context objects to oslo.policy directly\n\nThis oslo.policy Enforcer() object understands oslo.context\nRequestContext objects and will translate the appropriate policy values\nfor you if it\u0027s passed into enforcement.\n\nThis lowers the bar for people interacting with oslo.policy because they\ndon\u0027t have to translate important authorization inforamtion from\nRequestContext into a dictionary, specifically for policy enforcement.\nInstead, they can just pass in the context object and oslo.policy will\nknow what to do.\n\nThis commit updates the policy module to do just that and moves the\ncredential tranlations to where it\u0027s needed, for logging.\n\nChange-Id: I942da4b1880cf2312af2f8e15aef9f33be991f87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b7e364504f25554a9eb0eb7264b9a6c09408e8d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b7e364504f25554a9eb0eb7264b9a6c09408e8d6"}]},"branch":"refs/heads/master"},"70e5b229824c05939ac0398d89b72756709bb50c":{"kind":"REWORK","_number":2,"created":"2020-12-14 15:12:29.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/85/766785/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/85/766785/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/85/766785/2"}}},"commit":{"parents":[{"commit":"f0442937d9b91a1e0dc182a82520bab0b3cba7a1","subject":"Bump oslo.log version to 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f0442937d9b91a1e0dc182a82520bab0b3cba7a1"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-11 19:58:34.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-14 14:49:47.000000000","tz":0},"subject":"Pass context objects to oslo.policy directly","message":"Pass context objects to oslo.policy directly\n\nThis oslo.policy Enforcer() object understands oslo.context\nRequestContext objects and will translate the appropriate policy values\nfor you if it\u0027s passed into enforcement.\n\nThis lowers the bar for people interacting with oslo.policy because they\ndon\u0027t have to translate important authorization inforamtion from\nRequestContext into a dictionary, specifically for policy enforcement.\nInstead, they can just pass in the context object and oslo.policy will\nknow what to do. As a result, we need to make sure we\u0027re using newer\nversions of oslo.policy and oslo.context, hence the updates to\nlower-constraints.txt.\n\nThis commit updates the policy module to do just that and moves the\ncredential tranlations to where it\u0027s needed, for logging.\n\nChange-Id: I942da4b1880cf2312af2f8e15aef9f33be991f87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/70e5b229824c05939ac0398d89b72756709bb50c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/70e5b229824c05939ac0398d89b72756709bb50c"}]},"branch":"refs/heads/master"},"42b50300c7ce1e9ba77a25547a9c46cf0462bc0a":{"kind":"REWORK","_number":3,"created":"2020-12-14 20:05:07.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/85/766785/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/85/766785/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/85/766785/3"}}},"commit":{"parents":[{"commit":"4a3cf543e510393aec7eb670fedef0fadf1f3c2c","subject":"Bump oslo.log version to 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/4a3cf543e510393aec7eb670fedef0fadf1f3c2c"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-11 19:58:34.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-14 19:58:46.000000000","tz":0},"subject":"Pass context objects to oslo.policy directly","message":"Pass context objects to oslo.policy directly\n\nThis oslo.policy Enforcer() object understands oslo.context\nRequestContext objects and will translate the appropriate policy values\nfor you if it\u0027s passed into enforcement.\n\nThis lowers the bar for people interacting with oslo.policy because they\ndon\u0027t have to translate important authorization inforamtion from\nRequestContext into a dictionary, specifically for policy enforcement.\nInstead, they can just pass in the context object and oslo.policy will\nknow what to do. As a result, we need to make sure we\u0027re using newer\nversions of oslo.policy and oslo.context, hence the updates to\nlower-constraints.txt.\n\nThis commit updates the policy module to do just that and moves the\ncredential tranlations to where it\u0027s needed, for logging.\n\nChange-Id: I942da4b1880cf2312af2f8e15aef9f33be991f87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/42b50300c7ce1e9ba77a25547a9c46cf0462bc0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/42b50300c7ce1e9ba77a25547a9c46cf0462bc0a"}]},"branch":"refs/heads/master"},"0e66d5550e4e06b6ee1e8f2290bada79fb5bd641":{"kind":"REWORK","_number":4,"created":"2020-12-15 00:24:22.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/85/766785/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/85/766785/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/85/766785/4"}}},"commit":{"parents":[{"commit":"4a3cf543e510393aec7eb670fedef0fadf1f3c2c","subject":"Bump oslo.log version to 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/4a3cf543e510393aec7eb670fedef0fadf1f3c2c"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-11 19:58:34.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-15 00:18:26.000000000","tz":0},"subject":"Pass context objects to oslo.policy directly","message":"Pass context objects to oslo.policy directly\n\nThis oslo.policy Enforcer() object understands oslo.context\nRequestContext objects and will translate the appropriate policy values\nfor you if it\u0027s passed into enforcement.\n\nThis lowers the bar for people interacting with oslo.policy because they\ndon\u0027t have to translate important authorization inforamtion from\nRequestContext into a dictionary, specifically for policy enforcement.\nInstead, they can just pass in the context object and oslo.policy will\nknow what to do. As a result, we need to make sure we\u0027re using newer\nversions of oslo.policy and oslo.context, hence the updates to\nlower-constraints.txt.\n\nThis commit updates the policy module to do just that and moves the\ncredential tranlations to where it\u0027s needed, for logging.\n\nChange-Id: I942da4b1880cf2312af2f8e15aef9f33be991f87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/0e66d5550e4e06b6ee1e8f2290bada79fb5bd641"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/0e66d5550e4e06b6ee1e8f2290bada79fb5bd641"}]},"branch":"refs/heads/master"},"89d97870a08fcf9b658d3297b1f697f13b94bcc7":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-12-16 13:18:47.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/85/766785/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/85/766785/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/85/766785/5"}}},"commit":{"parents":[{"commit":"bf751919f0343aeb1193e695453bf60263c4f6d5","subject":"Bump oslo.log version to 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/bf751919f0343aeb1193e695453bf60263c4f6d5"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-11 19:58:34.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-16 13:12:32.000000000","tz":0},"subject":"Pass context objects to oslo.policy directly","message":"Pass context objects to oslo.policy directly\n\nThis oslo.policy Enforcer() object understands oslo.context\nRequestContext objects and will translate the appropriate policy values\nfor you if it\u0027s passed into enforcement.\n\nThis lowers the bar for people interacting with oslo.policy because they\ndon\u0027t have to translate important authorization inforamtion from\nRequestContext into a dictionary, specifically for policy enforcement.\nInstead, they can just pass in the context object and oslo.policy will\nknow what to do. As a result, we need to make sure we\u0027re using newer\nversions of oslo.policy and oslo.context, hence the updates to\nlower-constraints.txt.\n\nThis commit updates the policy module to do just that and moves the\ncredential tranlations to where it\u0027s needed, for logging.\n\nChange-Id: I942da4b1880cf2312af2f8e15aef9f33be991f87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/89d97870a08fcf9b658d3297b1f697f13b94bcc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/89d97870a08fcf9b658d3297b1f697f13b94bcc7"}]},"branch":"refs/heads/master"},"dc4fb83cfc45a22d0be7bcac212adc5cf349dd3b":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-12-17 14:53:52.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/85/766785/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/85/766785/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/85/766785/6"}}},"commit":{"parents":[{"commit":"793b2534739cd45db652375a5cc565361a01308b","subject":"Bump oslo.log version to 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/793b2534739cd45db652375a5cc565361a01308b"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-11 19:58:34.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-17 14:46:52.000000000","tz":0},"subject":"Pass context objects to oslo.policy directly","message":"Pass context objects to oslo.policy directly\n\nThis oslo.policy Enforcer() object understands oslo.context\nRequestContext objects and will translate the appropriate policy values\nfor you if it\u0027s passed into enforcement.\n\nThis lowers the bar for people interacting with oslo.policy because they\ndon\u0027t have to translate important authorization inforamtion from\nRequestContext into a dictionary, specifically for policy enforcement.\nInstead, they can just pass in the context object and oslo.policy will\nknow what to do. As a result, we need to make sure we\u0027re using newer\nversions of oslo.policy and oslo.context, hence the updates to\nlower-constraints.txt.\n\nThis commit updates the policy module to do just that and moves the\ncredential tranlations to where it\u0027s needed, for logging.\n\nChange-Id: I942da4b1880cf2312af2f8e15aef9f33be991f87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/dc4fb83cfc45a22d0be7bcac212adc5cf349dd3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/dc4fb83cfc45a22d0be7bcac212adc5cf349dd3b"}]},"branch":"refs/heads/master"},"32ef0c08247803f9b9e4ff7368c66cff99b3216a":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-12-17 23:07:51.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/85/766785/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/85/766785/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/85/766785/7"}}},"commit":{"parents":[{"commit":"c9cba351ecfdca08023098c7cf5cc150aa5775e6","subject":"Bump oslo.log version to 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/c9cba351ecfdca08023098c7cf5cc150aa5775e6"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-11 19:58:34.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-17 23:01:34.000000000","tz":0},"subject":"Pass context objects to oslo.policy directly","message":"Pass context objects to oslo.policy directly\n\nThis oslo.policy Enforcer() object understands oslo.context\nRequestContext objects and will translate the appropriate policy values\nfor you if it\u0027s passed into enforcement.\n\nThis lowers the bar for people interacting with oslo.policy because they\ndon\u0027t have to translate important authorization inforamtion from\nRequestContext into a dictionary, specifically for policy enforcement.\nInstead, they can just pass in the context object and oslo.policy will\nknow what to do. As a result, we need to make sure we\u0027re using newer\nversions of oslo.policy and oslo.context, hence the updates to\nlower-constraints.txt.\n\nThis commit updates the policy module to do just that and moves the\ncredential tranlations to where it\u0027s needed, for logging.\n\nChange-Id: I942da4b1880cf2312af2f8e15aef9f33be991f87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/32ef0c08247803f9b9e4ff7368c66cff99b3216a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/32ef0c08247803f9b9e4ff7368c66cff99b3216a"}]},"branch":"refs/heads/master"},"fb088779dda7224584971049717aa61aef5228b7":{"kind":"REWORK","_number":8,"created":"2021-01-07 04:46:24.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/85/766785/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/85/766785/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/85/766785/8"}}},"commit":{"parents":[{"commit":"ed29666ffcf26668a1afb5dd909de399bdc6584c","subject":"Bump oslo.log version to 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ed29666ffcf26668a1afb5dd909de399bdc6584c"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-11 19:58:34.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-01-07 04:31:30.000000000","tz":0},"subject":"Pass context objects to oslo.policy directly","message":"Pass context objects to oslo.policy directly\n\nThis oslo.policy Enforcer() object understands oslo.context\nRequestContext objects and will translate the appropriate policy values\nfor you if it\u0027s passed into enforcement.\n\nThis lowers the bar for people interacting with oslo.policy because they\ndon\u0027t have to translate important authorization inforamtion from\nRequestContext into a dictionary, specifically for policy enforcement.\nInstead, they can just pass in the context object and oslo.policy will\nknow what to do. As a result, we need to make sure we\u0027re using newer\nversions of oslo.policy and oslo.context, hence the updates to\nlower-constraints.txt.\n\nThis commit updates the policy module to do just that and moves the\ncredential tranlations to where it\u0027s needed, for logging.\n\nChange-Id: I942da4b1880cf2312af2f8e15aef9f33be991f87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/fb088779dda7224584971049717aa61aef5228b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/fb088779dda7224584971049717aa61aef5228b7"}]},"branch":"refs/heads/master"},"e34420648763640ae668f3c694dd98df51eca1fc":{"kind":"REWORK","_number":9,"created":"2021-01-21 02:26:02.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/85/766785/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/85/766785/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/85/766785/9"}}},"commit":{"parents":[{"commit":"fa95ef8772eccd0d994b4e26668d84f2976931d5","subject":"[goal] Deprecate the JSON formatted policy file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/fa95ef8772eccd0d994b4e26668d84f2976931d5"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-11 19:58:34.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-01-21 02:19:14.000000000","tz":0},"subject":"Pass context objects to oslo.policy directly","message":"Pass context objects to oslo.policy directly\n\nThis oslo.policy Enforcer() object understands oslo.context\nRequestContext objects and will translate the appropriate policy values\nfor you if it\u0027s passed into enforcement.\n\nThis lowers the bar for people interacting with oslo.policy because they\ndon\u0027t have to translate important authorization inforamtion from\nRequestContext into a dictionary, specifically for policy enforcement.\nInstead, they can just pass in the context object and oslo.policy will\nknow what to do.\n\nThis commit updates the policy module to do just that and moves the\ncredential tranlations to where it\u0027s needed, for logging.\n\nChange-Id: I942da4b1880cf2312af2f8e15aef9f33be991f87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e34420648763640ae668f3c694dd98df51eca1fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e34420648763640ae668f3c694dd98df51eca1fc"}]},"branch":"refs/heads/master"},"560de6a37f7bb9511be11b00c2adb2a03435360f":{"kind":"NO_CHANGE","_number":10,"created":"2021-01-27 17:35:11.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/85/766785/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/85/766785/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/766785/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/85/766785/10"}}},"commit":{"parents":[{"commit":"fa95ef8772eccd0d994b4e26668d84f2976931d5","subject":"[goal] Deprecate the JSON formatted policy file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/fa95ef8772eccd0d994b4e26668d84f2976931d5"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-11 19:58:34.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-01-27 16:49:11.000000000","tz":0},"subject":"Pass context objects to oslo.policy directly","message":"Pass context objects to oslo.policy directly\n\nThis oslo.policy Enforcer() object understands oslo.context\nRequestContext objects and will translate the appropriate policy values\nfor you if it\u0027s passed into enforcement.\n\nThis lowers the bar for people interacting with oslo.policy because they\ndon\u0027t have to translate important authorization inforamtion from\nRequestContext into a dictionary, specifically for policy enforcement.\nInstead, they can just pass in the context object and oslo.policy will\nknow what to do.\n\nThis commit updates the policy module to do just that and moves the\ncredential tranlations to where it\u0027s needed, for logging.\n\nChange-Id: I942da4b1880cf2312af2f8e15aef9f33be991f87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/560de6a37f7bb9511be11b00c2adb2a03435360f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/560de6a37f7bb9511be11b00c2adb2a03435360f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
