)]}'
{"id":"openstack%2Fplacement~777174","triplet_id":"openstack%2Fplacement~master~I4fbab60f04f85d70c703385d7b43457045e3661d","project":"openstack/placement","branch":"master","topic":"fix-traits","hashtags":[],"change_id":"I4fbab60f04f85d70c703385d7b43457045e3661d","subject":"Update traits in tests and requirements","status":"MERGED","created":"2021-02-23 17:41:14.000000000","updated":"2021-02-24 22:25:03.000000000","submitted":"2021-02-24 22:22:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777174-fix-traits","meta_rev_id":"b34fa1288f8cc9da979c47d1e4e4fab91b4aa33a","_number":777174,"virtual_id_number":777174,"owner":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-24 22:22:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"recommended":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2021-02-24 14:01:39.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2021-02-24 13:54:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2021-02-24 14:15:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2021-02-24 14:15:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-23 17:43:23.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"state":"REVIEWER"},{"updated":"2021-02-23 17:43:23.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"state":"REVIEWER"},{"updated":"2021-02-23 17:43:23.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2021-02-23 22:55:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-24 13:54:36.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-02-24 14:01:39.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2021-02-24 14:15:21.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"70251efaa26b177f518058efed0a649ff7c773db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2021-02-23 17:41:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de761357a2ee290ae1c9b1b1ece4d5e1903116c9","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2021-02-23 17:43:23.000000000","message":"Patch Set 1:\n\nI probably won\u0027t be available if this needs tweaks but I wanted to get it started. Please feel free to adopt it. I\u0027m making this change because the gabbi CI tests against placement, so when placement is failing gabbi is too.","accounts_in_message":[],"_revision_number":1},{"id":"3de63fd84cacb3e2afff71002fd0e2ad138a711d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 22:55:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9b3bd5ac8200437da4e9134c556d2a99 : SUCCESS in 4m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/874f1faf3b0b45a9a543c9a040d761de : SUCCESS in 1h 11m 38s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/f9866e20d68a45c3bc85f30f38b6c3bb : SUCCESS in 1h 10m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f998fafc3cdf461da9f747041f1f60ca : SUCCESS in 7m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4acf0f20d8684b72b1567c8a55f35154 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1caf02f1e6240dc8392a7964149fc40 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7676bafdc2dd42399c48a5db003c5b8d : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e741f76365f48f9ac0f7a9d4781ef70 : SUCCESS in 8m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63d971d7d5b447c189e680014d050de9 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e5b81ce30334c66974827c75fc9fb8e : SUCCESS in 8m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fc6d735bcb924e4ab5033dbc0e981bb0 : SUCCESS in 5m 48s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/8869a3e0f93340738ff8cd65709810a3 : SUCCESS in 8m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8931ae8ad4624b67b0c3ebd09e709666 : SUCCESS in 9m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bf1223e1a53f4f319163a4cbbe8db890 : SUCCESS in 6m 21s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2fef0d74b5d646d0bfd921b434c4ce7e : SUCCESS in 13m 30s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/ef62737d5a19497790a9498435bd2706 : SUCCESS in 38m 50s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/76f00ffbcabf44d5afe028d3a18774f9 : SUCCESS in 12m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ad8c157bf094162843cab876949a275 : SUCCESS in 1h 03m 19s","accounts_in_message":[],"_revision_number":1},{"id":"164ad028dd6258f5fd123d9a14989f69e82aaf96","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-24 13:54:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cb9fb17d92a0ced85587761bac5dccde3a8f5b34","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-02-24 14:01:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"45cb59048fe1db2c2c6afd9dcb06a13d4891ffaa","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-02-24 14:15:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks Chris!","accounts_in_message":[],"_revision_number":1},{"id":"7a1f2c9604abee3623fb14473b2b9e87f8d99452","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 14:15:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ffb6ac823396ebfb504fe42cc5bb4256baee4ec7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 22:22:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d62626ed2c3c4d50a9c12cc20dfed53b : SUCCESS in 6m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/6ca127e7ff8b4c8db3fd2581b6dad976 : SUCCESS in 1h 05m 21s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/e8908e2604684f939c48ce9ed1ce8e1c : SUCCESS in 1h 17m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/406a31b807164c9b9a16f588dfb76ace : SUCCESS in 10m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd9394e8507c42ab91e605ac367c42a4 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/300b23a574b7405b800e67db65b892d3 : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f832feb549134522a9dabb00ca97af9a : SUCCESS in 8m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a0ed17423e94268b241059f338b6cd3 : SUCCESS in 9m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/05b1a7bea578485a80df69be72ecd472 : SUCCESS in 8m 03s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/13923ee3038a455ba7b3f30c31ba71cf : SUCCESS in 8m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f43c5d2cbabf4ad78f70c8635b9db69f : SUCCESS in 7m 58s\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0c93e5bf7f7a45dcb7c5697cee0bef52 : SUCCESS in 18m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1aef5d238f224a39ae54b1d77189cf7f : SUCCESS in 1h 01m 34s","accounts_in_message":[],"_revision_number":1},{"id":"a23cd4286cc404bbdac1d153ab9b51eb38ad3d54","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 22:22:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b34fa1288f8cc9da979c47d1e4e4fab91b4aa33a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 22:25:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7870654c55b43f28c92c1836f7985ea : SUCCESS in 1m 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f41599e1ada2dfc7900817b09b623422d9d1dbe9","revisions":{"f41599e1ada2dfc7900817b09b623422d9d1dbe9":{"kind":"REWORK","_number":1,"created":"2021-02-23 17:41:14.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/74/777174/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/74/777174/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/74/777174/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/74/777174/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/74/777174/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/74/777174/1"}}},"commit":{"parents":[{"commit":"b2ecae242d98931af9b466c6a75cb82c3e4bcb2b","subject":"policy: Add releasenote for RBAC work","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b2ecae242d98931af9b466c6a75cb82c3e4bcb2b"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2021-02-23 17:35:56.000000000","tz":0},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2021-02-23 17:41:12.000000000","tz":0},"subject":"Update traits in tests and requirements","message":"Update traits in tests and requirements\n\nIn I0438e39a3c61253f781761a5aae895993f897adb,\nI333147dcd47c6d0b926338a5a0c545f5adc63961 , and\nIa4d1871b719037174506fb784b77572e63303762 six new\ntraits were added, and a new os-traits was released\ntriggering the canary test for updating constraints.\n\nChange-Id: I4fbab60f04f85d70c703385d7b43457045e3661d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f41599e1ada2dfc7900817b09b623422d9d1dbe9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f41599e1ada2dfc7900817b09b623422d9d1dbe9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
