)]}'
{"id":"openstack%2Fplacement~783235","triplet_id":"openstack%2Fplacement~master~Ie137d670fb778b80160cc190de388af4243fe0db","project":"openstack/placement","branch":"master","topic":"add-xena-python-jobtemplates","hashtags":[],"change_id":"Ie137d670fb778b80160cc190de388af4243fe0db","subject":"Add Python3 xena unit tests","status":"MERGED","created":"2021-03-26 09:35:52.000000000","updated":"2021-03-26 14:31:19.000000000","submitted":"2021-03-26 14:23:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783235-add-xena-python-jobtemplates","meta_rev_id":"d626c5e50585bf27276a3fab39e02c7fd82bed95","_number":783235,"virtual_id_number":783235,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-26 14:22:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-03-26 09:39:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-26 10:07:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-26 10:07:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-26 09:39:15.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-03-26 10:07:02.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2021-03-26 11:31:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7e64d501d3c3698ca6f13f0580397444ed914aa6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-03-26 09:35:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6f601074e5c7c9019ec37979dc6c6edaedb49d4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-03-26 09:39:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c85f4aae092794953f812318dc95da24d66f4d83","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2021-03-26 10:07:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6c5dc52179a9624b39772121a5af0ef93f4935f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 11:31:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/40c2a3594d5c446b888b63979fbb0cd2 : SUCCESS in 1h 10m 03s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/10da7eddec1a433f955bd92970165de5 : SUCCESS in 1h 13m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/137e1527ea5944048e59f517a378c4dc : SUCCESS in 7m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9220016280741549c7f56242ee25dee : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80f94917dbad424195215ff4e6a3825a : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/016d200b9832468fbd18cd8f5063c5eb : SUCCESS in 7m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db78deb41bdd424a93a9155802545fb1 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4efb5c0c7594e579a054ff50958caf8 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa6117594cf04815af0548b44d457460 : SUCCESS in 9m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d49d19809dc84e5c95fb15872c0b05ce : SUCCESS in 6m 35s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/e85e8baaca5345cdb5f81828b15a046e : SUCCESS in 6m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a26683cbe97d4336ad204e12e0392b60 : SUCCESS in 5m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/22c7b461332e4d24bb07ef103b548524 : SUCCESS in 8m 35s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b9b5b8fd8ea2452fb5af186ecc342edf : SUCCESS in 16m 07s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/77e5cfaf64f0437c96b3a7f7becab331 : SUCCESS in 23m 12s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/7175cb9098da43e78cad0ebec7b72cc6 : SUCCESS in 8m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22e1b0bcf2394af5be788e301676715f : SUCCESS in 55m 33s","accounts_in_message":[],"_revision_number":1},{"id":"b371f0465be0d3a65def280d73deeea1baa7047e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 11:45:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e6193adef06a25b80074f52f86068a88f77fe1c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 14:22:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/8db3f59a9e6d46108ea8d950ba8fe433 : SUCCESS in 1h 04m 09s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/a952100950434d918a3727bada2ff38e : SUCCESS in 1h 17m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5969b75fd04b4effb21f1a534dbae728 : SUCCESS in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efd07c70ede547a4b0298e39dd8563ec : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49d23d5593754950980f159a36266b39 : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2259023f3e1e43009145d7cdb3254b7e : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e50a26276a03408bb6bcf237dce632c3 : SUCCESS in 10m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/106c86c3e2e44ca98c4bdaf0fea08923 : SUCCESS in 7m 49s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/46fd5636e63543aea76449e0dc43c89a : SUCCESS in 8m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7a284e348e2e4570ab654ab6a8785035 : SUCCESS in 8m 52s\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/22bda1a803e14cd08ea7dbaad2d53311 : SUCCESS in 14m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ee46cfecbee4070bf10e007aae59c8d : SUCCESS in 54m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e9aa500217347bede69b0ba803b6abd32514b05b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 14:23:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d626c5e50585bf27276a3fab39e02c7fd82bed95","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 14:31:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b74dbee9aaec4959b096db4af1450837 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6f00ba5f685183539d0ebf62a4741f2f6930e051","revisions":{"6f00ba5f685183539d0ebf62a4741f2f6930e051":{"kind":"REWORK","_number":1,"created":"2021-03-26 09:35:52.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/35/783235/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/35/783235/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/35/783235/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/35/783235/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/35/783235/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/35/783235/1"}}},"commit":{"parents":[{"commit":"48b9d6f2d7afb6f0b6d82cf60722bb4cd3494b03","subject":"Update master for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/48b9d6f2d7afb6f0b6d82cf60722bb4cd3494b03"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-26 09:35:49.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-26 09:35:49.000000000","tz":0},"subject":"Add Python3 xena unit tests","message":"Add Python3 xena unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for xena.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: Ie137d670fb778b80160cc190de388af4243fe0db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/6f00ba5f685183539d0ebf62a4741f2f6930e051"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/6f00ba5f685183539d0ebf62a4741f2f6930e051"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
