)]}'
{"id":"openstack%2Fplacement~799601","triplet_id":"openstack%2Fplacement~master~I6a2e22840060da9c1156581c1e1cea880d6ec840","project":"openstack/placement","branch":"master","topic":"update-min-tox","hashtags":[],"change_id":"I6a2e22840060da9c1156581c1e1cea880d6ec840","subject":"Changed minversion in tox to 3.18.0","status":"ABANDONED","created":"2021-07-06 08:15:05.000000000","updated":"2021-07-12 00:55:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"654a41dae5d4fe8e7cbfe4c78a7c4cbd37b955d7","_number":799601,"virtual_id_number":799601,"owner":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-07-06 08:32:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"all":[{"value":-1,"date":"2021-07-10 14:54:39.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-06 08:32:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-10 14:54:39.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"}],"messages":[{"id":"11dff22cfac65fc69b39a87afff65d7abd543b6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-07-06 08:15:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6a29b383be806a745689192afbe6d5ef3adde70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 08:32:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/07de0e81db344bcaa55fc1569cf33f2d : SUCCESS in 7m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4e5b9d223e24d299f6613897fcd4889 : SUCCESS in 8m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80900daafb5c45b28d67c63536b472bb : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bec62e0ae2a4aa3aee8f860f5c92c64 : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22b4d7cb5e034ab891c86b52f3007b04 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91740e96eb52491c8e36c66ed17d5b4f : SUCCESS in 7m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0563676cb3a64644bd65b424029309ef : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/344da6e973944d6ea7665fe61bc2b1a9 : SUCCESS in 9m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2598be95909645a9af264239ae21bc63 : SUCCESS in 6m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/55b0e8a2d7c94c8f94b79c0d98f803d1 : SUCCESS in 8m 26s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/adbdc0acb8914d33a03ec7ea23ea3c94 : SUCCESS in 8m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3658b7674fc842bc9faeffdeb3bc01d1 : SUCCESS in 7m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7c24b76832094db38c404031fcc2351d : SUCCESS in 7m 36s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8cbf4c8a420e4335b63f3368427181fb : SUCCESS in 15m 40s","accounts_in_message":[],"_revision_number":1},{"id":"94a52dc9ea7a5065e0b8fbdf546db3cbc5560d6c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"date":"2021-07-06 10:32:27.000000000","message":"Topic set to update-min-tox","accounts_in_message":[],"_revision_number":1},{"id":"5f4886addabcf9ef59a473222002d32357726c91","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2021-07-10 14:54:39.000000000","message":"Patch Set 1: Code-Review-1\n\nThis patch is a duplicate of the following patch:\n\nhttps://review.opendev.org/c/openstack/placement/+/791971","accounts_in_message":[],"_revision_number":1},{"id":"654a41dae5d4fe8e7cbfe4c78a7c4cbd37b955d7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-07-12 00:55:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"05839cc215949940f553d18ac8d738ff033822ce","revisions":{"05839cc215949940f553d18ac8d738ff033822ce":{"kind":"REWORK","_number":1,"created":"2021-07-06 08:15:05.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/01/799601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/01/799601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/01/799601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/01/799601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/01/799601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/01/799601/1"}}},"commit":{"parents":[{"commit":"f77a7f9928d1156450c48045c48597b2feec9cc1","subject":"Bump os-resource-classes deps to 1.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f77a7f9928d1156450c48045c48597b2feec9cc1"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2021-07-06 08:14:55.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2021-07-06 08:14:55.000000000","tz":480},"subject":"Changed minversion in tox to 3.18.0","message":"Changed minversion in tox to 3.18.0\n\nThe patch bumps min version of tox to 3.18.0 in order to\nreplace tox\u0027s whitelist_externals by allowlist_externals option:\nhttps://github.com/tox-dev/tox/blob/master/docs/changelog.rst#v3180-2020-07-23\n\nChange-Id: I6a2e22840060da9c1156581c1e1cea880d6ec840\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/05839cc215949940f553d18ac8d738ff033822ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/05839cc215949940f553d18ac8d738ff033822ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
