)]}'
{"id":"openstack%2Fplacement~800611","triplet_id":"openstack%2Fplacement~master~I4d15201dd2f94d00f4f2b53f773511ea020ef4ac","project":"openstack/placement","branch":"master","topic":"story/2009049","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-11-29 21:54:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4d15201dd2f94d00f4f2b53f773511ea020ef4ac","subject":"Enable HTTPProxyToWSGI middleware to find actual client ips","status":"MERGED","created":"2021-07-13 09:49:28.000000000","updated":"2021-11-29 21:56:53.000000000","submitted":"2021-11-29 21:54:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"800611-story/2009049","meta_rev_id":"909e39fc966a5528ff31683bfa216ee309414c89","_number":800611,"virtual_id_number":800611,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-29 21:54:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"recommended":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"all":[{"value":1,"date":"2021-08-19 13:27:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-11-29 17:02:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-11-29 17:28:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-11-29 17:28:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-13 10:59:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-30 16:57:24.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-07-31 00:54:06.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2021-11-29 17:02:47.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-11-29 17:28:53.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"}],"messages":[{"id":"a9be352562c33c1e6ddfe8e2a75eed816858371b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-13 09:49:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1105197768087eec11d237738e6b9fb765ce39d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-13 09:50:11.000000000","message":"Topic set to task/42819","accounts_in_message":[],"_revision_number":1},{"id":"1e0abff03bd02f1a38233d2172d10cffa1cb808b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-13 09:50:20.000000000","message":"Topic task/42819 removed","accounts_in_message":[],"_revision_number":1},{"id":"7ca095c965c77fe0def68f2c11fcb6b3493d04ea","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-13 09:50:25.000000000","message":"Topic set to story/2009049","accounts_in_message":[],"_revision_number":1},{"id":"fb1114c2e2cfe548b584bdaba705334a99e760c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-13 09:51:23.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e046a9a61605a370857064ec86a5da4df5c531e9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-13 09:51:59.000000000","message":"Topic task/42819 removed","accounts_in_message":[],"_revision_number":2},{"id":"ade7c7defcc3cae4fbf0f41ec8e6a82793b67091","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-13 09:52:02.000000000","message":"Topic set to story/2009049","accounts_in_message":[],"_revision_number":2},{"id":"93a73a28254965032b45ab343ad232e44aff416b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-13 09:53:04.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"54cb65f6a3696194d95904bc30678a25e8cfda55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-13 09:57:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d6e5aa2c301df71c1542f1bd5ac52f87b9d982b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-13 10:00:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"16619a7dd3260630ad257523d2af62fc4c94046e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 10:59:56.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/d67ca260b92d484c97533775bf220368 : FAILURE in 59m 05s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/c01d39b52cc14023bbd8a80f2a446948 : FAILURE in 21m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/536f44e382854d9792d3bec67c2a416f : FAILURE in 6m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f595ac1765ff41128c3b2064856078db : FAILURE in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb782f44fd2c4afaaa2b6bac3122524f : FAILURE in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ac0fab1e0af40bab52fa03384f62466 : FAILURE in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b05c2934a52d4c358d52361c19d264b6 : FAILURE in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58fd3e9de512411a8e29a29a2b132424 : FAILURE in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e12ac7955d448ca93c70c1f52463834 : SUCCESS in 13m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/065546dce67340749791a17cc455d1ca : FAILURE in 5m 54s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/f61b1ee8972441a2a397ae01041e06da : FAILURE in 6m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/75be310cb2144278996de86e036e73f1 : FAILURE in 6m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f30d63eff79941dab680faba8cb41595 : FAILURE in 6m 46s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/197a304bf7774a518733c4aa57674391 : FAILURE in 7m 18s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/cf0555629749461893702157606fbbcd : FAILURE in 5m 08s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/b17da390e6214a3a9e9e804ff6b9aaf0 : FAILURE in 3m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a11ce961ac2c4d8f9bfb8d1c87626211 : FAILURE in 28m 15s","accounts_in_message":[],"_revision_number":5},{"id":"80b2ec2d3f51ab848833cd4cbc97d0d9b0edb2fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-13 11:21:53.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"e5c325ad0bde5cea6f88f446295439a61018649b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 12:35:57.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/145e00356885484fa650bae7f6089e3d : SUCCESS in 52m 41s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/528f40a559964ebb890a9e13b7b961fa : SUCCESS in 1h 12m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59541fb08b8f47bf988bab01ee76d010 : SUCCESS in 7m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5316d361c1cd416bbdb325efbabc7217 : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8a7f9acbef44727b9181b522faf6367 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da614f1213ba4a1089751176386c6425 : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f4065c91fd34bd4b0474f2095cb47c2 : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da89b4c503e541b99006d75c54f472cb : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22adedf89d09422b91f45adb7e6e4432 : SUCCESS in 9m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/66bcefc7939f41068476ce359c5f130b : SUCCESS in 6m 26s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/f1a27d9e396c4d69aa12c5a9af81d5af : SUCCESS in 7m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/428c7147ceca4fc9be3b581777183f61 : SUCCESS in 6m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/043afd550b254f26b27bd93ecb00e27e : SUCCESS in 7m 18s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3bd33a7495e546f0886207df9671b8bf : SUCCESS in 16m 12s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/5fc5f13a7ce548b58ff8c2b1ead1933d : SUCCESS in 17m 51s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/322663ff81554d32b3680943eb2b57b9 : SUCCESS in 9m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db50768e86624d4cb3356228ef7dab94 : SUCCESS in 41m 26s","accounts_in_message":[],"_revision_number":6},{"id":"6674ffb63e0712ef970c7fa0ff693f729551db01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-13 15:20:11.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"179cd620d6a55bcd22790cd0a65034fafdfb42c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 16:25:41.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/7bbf85205065475c92fc6a7dfbe5ef4f : SUCCESS in 58m 29s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/cb9e46c3261441a3b7314d20c9cd79ca : SUCCESS in 1h 02m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce0f918c592049b0977486b2d67d563d : SUCCESS in 7m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/337e7bb08e514cdb92c2de383c11a332 : SUCCESS in 7m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5657d2a484a14008bde23fdb1374fb2a : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8781181dc2484cb5964eb4d6b1a52e0b : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a45577c6c6af4c809f0e9fe29c6c357d : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd75288647c544c88856f8f3908c30f2 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a50fde6f36947aba805a9f58386c28e : SUCCESS in 7m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f70baaf90f0f44fa8c934e4a21d57235 : SUCCESS in 5m 41s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/0e6689fc7ad642edafac5aaae5c587bd : SUCCESS in 7m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1d35f74d4bf24bc5af762f6a35f55119 : SUCCESS in 7m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ff9395c0055040398dac0d5f61832833 : SUCCESS in 6m 25s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ca6c3a0e8d6044d284923c7c818383ac : SUCCESS in 15m 36s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/0b4f75754b7a449d884b490e4399cff2 : SUCCESS in 18m 46s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/c3512fc7b0ff42a59b8e03f44d54502a : SUCCESS in 9m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6da4df059a984cb6a63875c07e225225 : SUCCESS in 57m 01s","accounts_in_message":[],"_revision_number":7},{"id":"4180a4c57ece37858832cb99c4082ea978f22d31","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-30 16:57:24.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"ffd28f848347a9557b190301a48f0e61c54038e1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-30 16:57:48.000000000","message":"Patch Set 7:\n\nCould potentially merit a release note. Will re-review if you can add one","accounts_in_message":[],"_revision_number":7},{"id":"dccbe612312aca6470ea8a63514bc998cc99cea6","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2021-07-31 00:54:06.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"75130a3420cd5ffb111d886760435ceb5894803d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-01 11:55:59.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"8c78ed08105fba10a2f01201280b1e628480699c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-01 11:56:53.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"920a366b4b5d17f0777f9a35175a471ca2a02ab6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-01 13:12:21.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/d7f9dc6c307a4a469371d14b3a60dcbc : SUCCESS in 1h 09m 29s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/cc528a6538db43159deac8938ee163c0 : SUCCESS in 1h 10m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cac1216d4de049ef93df33267613cc07 : SUCCESS in 6m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/649c68acca074086b2eaf9be7eeff440 : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dfd3d633eb34564a9d2e637d5253ae6 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3496378f88ab4b38b9a118aae926a3d0 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36e4dcf676b84ed68c0b599ffd11ff16 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d70c8cb790f446308e76537d6dead8de : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c4542eaad7c40c1b28ee9e59d036acf : SUCCESS in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c606801ce9674405ab9ac3babb161550 : SUCCESS in 5m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9f60b72f6e1745699967b18bc1d7bf3a : SUCCESS in 5m 42s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/ae938efa9d7146ab86d74564dffc0575 : SUCCESS in 5m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/02788347597f43d48e6469ef22e399a3 : SUCCESS in 5m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a17b392d50044241b2044fcf7efdfc7a : SUCCESS in 6m 29s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7968d1a1fc094150a158c5b39f5021c4 : SUCCESS in 13m 47s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/05aead6ca4d14396b74ae9455eacf578 : SUCCESS in 20m 01s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/c609adf95ac045d1b50e6ba72a284f82 : SUCCESS in 9m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b121df35586a415191cdefc1babfb506 : SUCCESS in 48m 18s","accounts_in_message":[],"_revision_number":9},{"id":"e3654db753b74564302651df64cb3bd91c0489a3","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2021-08-19 13:27:57.000000000","message":"Patch Set 9: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":9},{"id":"f3e5490225a7d1a24ef4f2d8e7fbd88e81a28153","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-11-29 17:02:47.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"f7fbc488be42dbe9afcbb1aab3a3c6c2d2242583","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2021-11-29 17:28:53.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"e28b819d45d6506867706b1e3d88b73dc31befb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 17:30:03.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"c55670e2bff1c099b37da0bb018149782c8615fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 21:54:55.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/da6af0477c9e4de586261c86432ef73c : SUCCESS in 1h 05m 40s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/3f91999ee80a4b9987585f45cffd39b3 : SUCCESS in 1h 12m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d36544c863dd4bba81103f89fb3ecaee : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/210ebc7ab9694718853aa818053f3f53 : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd9409ace9154b84a7a83fe21158f704 : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5f2b65798d04726beb8e42e8455abda : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f20be9998b45418a823b6daf955ed3d8 : SUCCESS in 7m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db9ca1d6034b4647bbda425bfb07a803 : SUCCESS in 6m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f0933894d2a14bb68aff40a7738a1cc0 : SUCCESS in 5m 57s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/112d9dce7c5b4dd99f4357cf473bf836 : SUCCESS in 6m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/20eb21152e9c47d9a30f9bda4a7cd951 : SUCCESS in 7m 01s\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9bce3afd7c0441a89c346ca723a6f83d : SUCCESS in 18m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/482f53105792494cabde388cfa20672e : SUCCESS in 48m 52s","accounts_in_message":[],"_revision_number":9},{"id":"7262b32e7e19ca1450ee3be64bee1c62f5e6069a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 21:54:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"909e39fc966a5528ff31683bfa216ee309414c89","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 21:56:53.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2597e3ca1bc4fb4861f3e7718698c61 : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63020c57d8644f19b5a3f87d7b6a456f : SUCCESS in 54s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"2007f8fb458c7e333d7dfef2159b8de153583ced","revisions":{"76d2000ef7d35a431c53d0c9af2e7562e4eb3976":{"kind":"REWORK","_number":1,"created":"2021-07-13 09:49:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/800611/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/11/800611/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/11/800611/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/11/800611/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/11/800611/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/11/800611/1"}}},"commit":{"parents":[{"commit":"f77a7f9928d1156450c48045c48597b2feec9cc1","subject":"Bump os-resource-classes deps to 1.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f77a7f9928d1156450c48045c48597b2feec9cc1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-13 09:45:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-13 09:49:25.000000000","tz":540},"subject":"Enable HTTPProxyToWSGI to look up actual remote address","message":"Enable HTTPProxyToWSGI to look up actual remote address\n\nCurrently placement records REMOTE_ADDR environment as source ip but\nthis environment points not load balancer ips instead of actual client\nips when placmenet runs behind load balancers like haproxy.\n\nThis change enables HTTPProxyToWSGI middleware to parse forward\nheader to look up actual client address.\n\nStory: 2009049\nTask: 42819\nChange-Id: I4d15201dd2f94d00f4f2b53f773511ea020ef4ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/76d2000ef7d35a431c53d0c9af2e7562e4eb3976"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/76d2000ef7d35a431c53d0c9af2e7562e4eb3976"}]},"branch":"refs/heads/master"},"f648b091dd70f99e039e707670043e7295bf74d6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-07-13 09:51:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/800611/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/11/800611/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/11/800611/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/11/800611/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/11/800611/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/11/800611/2"}}},"commit":{"parents":[{"commit":"f77a7f9928d1156450c48045c48597b2feec9cc1","subject":"Bump os-resource-classes deps to 1.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f77a7f9928d1156450c48045c48597b2feec9cc1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-13 09:45:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-13 09:51:02.000000000","tz":540},"subject":"Enable HTTPProxyToWSGI to look up actual remote address","message":"Enable HTTPProxyToWSGI to look up actual remote address\n\nCurrently placement records REMOTE_ADDR environment as source ip but\nthis environment points load balancer ips instead of actual client ips\nwhen placmenet runs behind load balancers like haproxy.\n\nThis change enables HTTPProxyToWSGI middleware to parse forward\nheader to look up actual client address.\n\nStory: 2009049\nTask: 42819\nChange-Id: I4d15201dd2f94d00f4f2b53f773511ea020ef4ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f648b091dd70f99e039e707670043e7295bf74d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f648b091dd70f99e039e707670043e7295bf74d6"}]},"branch":"refs/heads/master"},"1b58151faf767385c905b35d768e6fa578070f2f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-07-13 09:53:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/800611/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/11/800611/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/11/800611/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/11/800611/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/11/800611/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/11/800611/3"}}},"commit":{"parents":[{"commit":"f77a7f9928d1156450c48045c48597b2feec9cc1","subject":"Bump os-resource-classes deps to 1.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f77a7f9928d1156450c48045c48597b2feec9cc1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-13 09:45:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-13 09:52:43.000000000","tz":540},"subject":"Enable HTTPProxyToWSGI middleware to find actual client ips","message":"Enable HTTPProxyToWSGI middleware to find actual client ips\n\nCurrently placement records REMOTE_ADDR environment as source ip but\nthis environment points load balancer ips instead of actual client ips\nwhen placmenet runs behind load balancers like haproxy.\n\nThis change enables HTTPProxyToWSGI middleware to parse request\nheaders to look up actual client address.\n\nStory: 2009049\nTask: 42819\nChange-Id: I4d15201dd2f94d00f4f2b53f773511ea020ef4ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/1b58151faf767385c905b35d768e6fa578070f2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/1b58151faf767385c905b35d768e6fa578070f2f"}]},"branch":"refs/heads/master"},"f04ac175be3f025263628290ee35cd3e1b91fd21":{"kind":"REWORK","_number":4,"created":"2021-07-13 09:57:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/800611/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/11/800611/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/11/800611/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/11/800611/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/11/800611/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/11/800611/4"}}},"commit":{"parents":[{"commit":"f77a7f9928d1156450c48045c48597b2feec9cc1","subject":"Bump os-resource-classes deps to 1.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f77a7f9928d1156450c48045c48597b2feec9cc1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-13 09:45:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-13 09:57:46.000000000","tz":540},"subject":"Enable HTTPProxyToWSGI middleware to find actual client ips","message":"Enable HTTPProxyToWSGI middleware to find actual client ips\n\nCurrently placement records REMOTE_ADDR environment as source ip but\nthis environment points load balancer ips instead of actual client ips\nwhen placmenet runs behind load balancers like haproxy.\n\nThis change enables HTTPProxyToWSGI middleware to parse request\nheaders to look up actual client address.\n\nStory: 2009049\nTask: 42819\nChange-Id: I4d15201dd2f94d00f4f2b53f773511ea020ef4ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f04ac175be3f025263628290ee35cd3e1b91fd21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f04ac175be3f025263628290ee35cd3e1b91fd21"}]},"branch":"refs/heads/master"},"5c3caa585c1b7cbea6166aed3c960c96fc907a96":{"kind":"REWORK","_number":5,"created":"2021-07-13 10:00:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/800611/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/11/800611/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/11/800611/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/11/800611/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/11/800611/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/11/800611/5"}}},"commit":{"parents":[{"commit":"f77a7f9928d1156450c48045c48597b2feec9cc1","subject":"Bump os-resource-classes deps to 1.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f77a7f9928d1156450c48045c48597b2feec9cc1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-13 09:45:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-13 10:00:20.000000000","tz":540},"subject":"Enable HTTPProxyToWSGI middleware to find actual client ips","message":"Enable HTTPProxyToWSGI middleware to find actual client ips\n\nCurrently placement records REMOTE_ADDR environment as source ip but\nthis environment points load balancer ips instead of actual client ips\nwhen placmenet runs behind load balancers like haproxy.\n\nThis change enables HTTPProxyToWSGI middleware to parse request\nheaders to look up actual client address.\n\nStory: 2009049\nTask: 42819\nChange-Id: I4d15201dd2f94d00f4f2b53f773511ea020ef4ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/5c3caa585c1b7cbea6166aed3c960c96fc907a96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/5c3caa585c1b7cbea6166aed3c960c96fc907a96"}]},"branch":"refs/heads/master"},"3096c2ffe050eb2709d8c808a3ca12b7ca605571":{"kind":"REWORK","_number":6,"created":"2021-07-13 11:21:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/800611/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/11/800611/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/11/800611/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/11/800611/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/11/800611/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/11/800611/6"}}},"commit":{"parents":[{"commit":"f77a7f9928d1156450c48045c48597b2feec9cc1","subject":"Bump os-resource-classes deps to 1.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f77a7f9928d1156450c48045c48597b2feec9cc1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-13 09:45:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-13 11:21:46.000000000","tz":0},"subject":"Enable HTTPProxyToWSGI middleware to find actual client ips","message":"Enable HTTPProxyToWSGI middleware to find actual client ips\n\nCurrently placement records REMOTE_ADDR environment as source ip but\nthis environment points load balancer ips instead of actual client ips\nwhen placmenet runs behind load balancers like haproxy.\n\nThis change enables HTTPProxyToWSGI middleware to parse request\nheaders to look up actual client address.\n\nStory: 2009049\nTask: 42819\nChange-Id: I4d15201dd2f94d00f4f2b53f773511ea020ef4ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/3096c2ffe050eb2709d8c808a3ca12b7ca605571"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/3096c2ffe050eb2709d8c808a3ca12b7ca605571"}]},"branch":"refs/heads/master"},"e60a2c6f784af4dbb22bcc324dcf4f43d70f7a14":{"kind":"REWORK","_number":7,"created":"2021-07-13 15:20:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/800611/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/11/800611/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/11/800611/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/11/800611/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/11/800611/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/11/800611/7"}}},"commit":{"parents":[{"commit":"f77a7f9928d1156450c48045c48597b2feec9cc1","subject":"Bump os-resource-classes deps to 1.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f77a7f9928d1156450c48045c48597b2feec9cc1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-13 09:45:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-13 15:20:04.000000000","tz":540},"subject":"Enable HTTPProxyToWSGI middleware to find actual client ips","message":"Enable HTTPProxyToWSGI middleware to find actual client ips\n\nCurrently placement records REMOTE_ADDR environment as source ip but\nthis environment points load balancer ips instead of actual client ips\nwhen placmenet runs behind load balancers like haproxy.\n\nThis change enables HTTPProxyToWSGI middleware to parse request\nheaders to look up actual client address.\n\nStory: 2009049\nTask: 42819\nChange-Id: I4d15201dd2f94d00f4f2b53f773511ea020ef4ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e60a2c6f784af4dbb22bcc324dcf4f43d70f7a14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e60a2c6f784af4dbb22bcc324dcf4f43d70f7a14"}]},"branch":"refs/heads/master"},"75f0317ffc6d3815ed085060ab1cae0aa751df46":{"kind":"REWORK","_number":8,"created":"2021-08-01 11:55:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/800611/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/11/800611/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/11/800611/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/11/800611/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/11/800611/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/11/800611/8"}}},"commit":{"parents":[{"commit":"f77a7f9928d1156450c48045c48597b2feec9cc1","subject":"Bump os-resource-classes deps to 1.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f77a7f9928d1156450c48045c48597b2feec9cc1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-13 09:45:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-01 11:55:37.000000000","tz":540},"subject":"Enable HTTPProxyToWSGI middleware to find actual client ips","message":"Enable HTTPProxyToWSGI middleware to find actual client ips\n\nCurrently placement records REMOTE_ADDR environment as source ip but\nthis environment points load balancer ips instead of actual client ips\nwhen placmenet runs behind load balancers like haproxy.\n\nThis change enables HTTPProxyToWSGI middleware to parse request\nheaders to look up actual client address.\n\nStory: 2009049\nTask: 42819\nChange-Id: I4d15201dd2f94d00f4f2b53f773511ea020ef4ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/75f0317ffc6d3815ed085060ab1cae0aa751df46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/75f0317ffc6d3815ed085060ab1cae0aa751df46"}]},"branch":"refs/heads/master"},"2007f8fb458c7e333d7dfef2159b8de153583ced":{"kind":"REWORK","_number":9,"created":"2021-08-01 11:56:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/800611/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/11/800611/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/11/800611/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/11/800611/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/11/800611/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/11/800611/9"}}},"commit":{"parents":[{"commit":"f77a7f9928d1156450c48045c48597b2feec9cc1","subject":"Bump os-resource-classes deps to 1.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f77a7f9928d1156450c48045c48597b2feec9cc1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-13 09:45:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-01 11:56:44.000000000","tz":540},"subject":"Enable HTTPProxyToWSGI middleware to find actual client ips","message":"Enable HTTPProxyToWSGI middleware to find actual client ips\n\nCurrently placement records REMOTE_ADDR environment as source ip but\nthis environment points load balancer ips instead of actual client ips\nwhen placmenet runs behind load balancers like haproxy.\n\nThis change enables HTTPProxyToWSGI middleware to parse request\nheaders to look up actual client address.\n\nStory: 2009049\nTask: 42819\nChange-Id: I4d15201dd2f94d00f4f2b53f773511ea020ef4ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/2007f8fb458c7e333d7dfef2159b8de153583ced"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/2007f8fb458c7e333d7dfef2159b8de153583ced"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
