)]}'
{"id":"openstack%2Fplacement~801105","triplet_id":"openstack%2Fplacement~master~Ie93f52777da62fa98e8b33d4d1ca0ae6da4de9e0","project":"openstack/placement","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-04-14 15:56:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie93f52777da62fa98e8b33d4d1ca0ae6da4de9e0","subject":"db: Replace deprecated \u0027FromClause.select().whereclause\u0027 parameter","status":"MERGED","created":"2021-07-16 16:50:39.000000000","updated":"2022-04-14 15:56:29.000000000","submitted":"2022-04-14 15:56:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"801105-sqlalchemy-20","meta_rev_id":"d6cfcfaf6e79f383d6efdbc4e2b8ea9de34eb34f","_number":801105,"virtual_id_number":801105,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-14 15:56:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2022-04-14 09:41:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-14 15:56:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2022-04-14 13:49:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2022-04-14 09:41:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2022-04-14 13:49:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-16 17:18:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-08 12:01:24.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2022-04-14 09:41:49.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2022-04-14 13:49:02.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"}],"messages":[{"id":"25ed40731bbb0d78379317df7975c61f9483fd41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-16 16:50:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9984583d2954480913556780923f4bbdcb7cc3a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 17:18:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/808a4fb91e414b5fa9004d47e5f32c85 : FAILURE in 7m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16159d18950a46d9973708a924b46df2 : FAILURE in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b519bbbdfac74674990e470292d2ae97 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de5c851dc1994b538143bc3e83e466fb : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/186cb76ff23842a09bc6a11f72665e1b : SUCCESS in 8m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c02e50e93738458086b8f69a9d429f76 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/045979a0e67a454dbb0fdf60ff41613d : SUCCESS in 10m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c412ef2995124ed3bcc84351f1777f3b : FAILURE in 6m 26s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/54b266371ad3440990e54d41ded48b1b : FAILURE in 8m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a0f19f727e3c4f3781492c635c6a25e8 : FAILURE in 6m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/40292730e7b84312ad1936bb9dcbc9f5 : FAILURE in 8m 06s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/502b4b44e4c24bce81281e54231083c1 : SUCCESS in 18m 38s","accounts_in_message":[],"_revision_number":1},{"id":"8f017597ed5484cebf35186de3bb858f6526a7b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-27 17:28:35.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"e086bf6cb49dded80340810f47b691bf056950af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 18:21:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f36f10536224fd394a708dfc97113e0 : SUCCESS in 7m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f30334e38ae641548cabcd6a9a0becc5 : FAILURE in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9186f3f2b5d941b0b41b441f97f1c4e6 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b9395c7627b40918da3d05f4636be9e : SUCCESS in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7439a40de394b18b31e9cea70a91a08 : SUCCESS in 7m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c977eca5425348d998b3269b48105b16 : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b15a13655d24c07b8136439d80b8768 : SUCCESS in 8m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/27a5cf3f61b146f3bcb8d579f05f286f : SUCCESS in 6m 44s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/6fa4635144a0402b9a3e7831dbf8bf21 : SUCCESS in 7m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d5a050748d2648939f47fbb0206faf1d : SUCCESS in 7m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a26688a29b314bb386329380916c40be : SUCCESS in 8m 08s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c7ee31017f224fefaffb1e52ecf2b039 : SUCCESS in 15m 18s","accounts_in_message":[],"_revision_number":2},{"id":"77a9c4df81bf4aff00587c370702fc7e1dea0304","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-28 09:44:06.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"905bb227b57f84049497df32f2e8f647fc4884db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 10:00:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cec43e2ebca04ad09b06461c0105eca8 : SUCCESS in 7m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f9ef65891ec4c55b81a9b8b3057fa8e : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/026f1c77640f4dccab922fc521fcdee1 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/500129cbf4b14c68bc275fba27b9a3cc : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc78d3880d474732af975342fcffa2bd : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7002d5ca0b8e4fa18ed828f9779c4076 : SUCCESS in 9m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54c29d8a64fc4a528cc86e48aabdf37b : SUCCESS in 8m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1de2215df2704c64a5a54aaa1d920d4f : SUCCESS in 6m 36s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/a03e4865679047c0a7971c49f62972dd : SUCCESS in 7m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/20b3054671ae404fbf8aa97b541b1d1a : SUCCESS in 6m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a7f26dc396674bff98ebd4930cb7f34d : SUCCESS in 6m 47s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2db81367de7045cdbee22f4833d3581c : SUCCESS in 14m 08s","accounts_in_message":[],"_revision_number":3},{"id":"606082dc3a12a94879c05d95001101e121433c66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-05 16:57:36.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"d54495ee1a41d1ce466cd851b00f816e9200dc20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-05 17:00:57.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"57d72d136b539eaf222f1cd5f88fd14b3163e0dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 17:56:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3a4348fdb754a0a91d6471196afd000 : SUCCESS in 48m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9d8de751f9f4ad491329e5b61b8987b : SUCCESS in 7m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64c917dfdab44b5fb595a2269bccb2b0 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e833d3315c5b4820bdeff3116fff524b : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7e20dea39614046986df4c82b4b3b2d : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/253125cfa3384e0a8df950db837940c0 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93f7d7d941e348c39803cedd3675f67f : SUCCESS in 9m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2217ae6692de463faa17f0339537530d : SUCCESS in 6m 04s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/823ca20a585c40b4b76def76d136625f : SUCCESS in 7m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b3b5b036135949518d6f75d6e0cdc033 : SUCCESS in 6m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ce60a4ad6d6f488aabb46cdddc6d7532 : SUCCESS in 6m 09s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/53abd35ee0334f5c8672bff7526831ef : SUCCESS in 17m 51s","accounts_in_message":[],"_revision_number":5},{"id":"05163fa1f46ab72882a392c2c2865d09ada9fc58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-02-07 12:29:58.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"daa76cf1272765fa2b4be7b460f9a0c5f1c4fffc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 13:18:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2fdad2ebf23e47c888e76f878ed4bcac : SUCCESS in 36m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe91aa7239fb4020b5b8b0f2e67aadfc : SUCCESS in 8m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6fdf027855c14c23b3a888d9c56e4f43 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5acfdc7cc26840caafc493b8e4fddec2 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c17e576c6a44f68babfc70449b98861 : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca09509e48b44bf5a200fb1711266144 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc3cbb1b1ebc4938920b500bd040e633 : SUCCESS in 9m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7d64f47d491c41c3a358d7c5478758cf : SUCCESS in 5m 08s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/86fbdf670f3145d087101930da80d8b4 : SUCCESS in 6m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7784bcef9493449cb02db775d1471210 : SUCCESS in 7m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ca4fac33ab0f44cc8efe01ca34256a46 : SUCCESS in 5m 01s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f96cdf9b982a4a04bda17d4e7a4836e9 : SUCCESS in 13m 10s","accounts_in_message":[],"_revision_number":6},{"id":"526376b05d3c2a470a4d28a0db5935f01e76b2ab","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-02-08 12:01:24.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"efa72deea42e79d63ace955dca0115c8fbbca7a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-03-24 15:54:52.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"627f73108026f0289c8c13914ec002785a927630","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 17:13:52.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/716fa136db0c416c90b9d57c3e536e45 : SUCCESS in 1h 09m 50s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d3b957c34fe3428da5d274d99489a43f : SUCCESS in 1h 02m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29e883fd2362482a873381ea73841ea5 : SUCCESS in 9m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/101cb483cbfd46c7995094f94ee913fe : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a1026d3ae4c4aa4a7cf6b8ce76744d1 : SUCCESS in 12m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a259381ab32545919105734c5b7f780c : SUCCESS in 7m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/077d0e9de71b4a3a845a7d781094652b : SUCCESS in 8m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d181b09f53c34ebf8229e021c513c036 : RETRY_LIMIT in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b46620285bd48f08d1f68c3cf93a2d5 : SUCCESS in 15m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/acc29e456e1c423e91880ad8f3d9c472 : SUCCESS in 7m 21s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/a36b7e2c53d2435f9091d1df40c95633 : SUCCESS in 7m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ee6c10ac731e438ca5cda44e8b0b780d : SUCCESS in 11m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e21bb5f1f4314de1ac1a9bb0fb16cecb : SUCCESS in 8m 21s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/efe6afc2eefc49bcac21e6173e42142b : SUCCESS in 20m 55s","accounts_in_message":[],"_revision_number":7},{"id":"3966177be8310aa3aa4a88cdb5463af3a72dc4a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-03-24 20:14:12.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"313b86deb77ee6b520e97f1c867a7ddf58d35169","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 21:12:41.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/cba423a8e71247fc852c671c78352281 : FAILURE in 56m 09s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f354bc18806346378d9ccef6bcf407eb : FAILURE in 46m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41fd24d9dfde41a283e4f1895726c5c6 : FAILURE in 8m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e634f5d92a2948f5ba09df82f8c752fa : FAILURE in 9m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f01914538b714fc38b16c7211f560b1c : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3919399ffd3e48fa81182e1f9881a4b8 : SUCCESS in 7m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/524a2d4a38e74cd5ae6f105c73612eaf : SUCCESS in 8m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/55b0c4dee7bf4481bb5883d3e8c37ff2 : RETRY_LIMIT in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64a59860dfc3467d839628ad284f79c0 : SUCCESS in 10m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/abdadf8860f643e28f89577271557239 : FAILURE in 9m 46s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/f093f38ceace4ee7bee92ef3f9d03280 : FAILURE in 9m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1f3bcf1b4dfa489b94cd9d9726b73cca : FAILURE in 7m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3b72d5b5f8a14271b9c6f79cdc9ae100 : FAILURE in 8m 57s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f740f525ed2a428998cf39f74d597114 : FAILURE in 24m 06s","accounts_in_message":[],"_revision_number":8},{"id":"79d0d6ecb573249a8622e5e238047f40cf1246ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-04-07 14:40:56.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"531b15f03ad7b8acc539bc8a65be65048196a278","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 15:25:15.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1c2b94d7b92741cc9ddfc2a4a501bf34 : SUCCESS in 40m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1af236cb461b4ac4a7832cfab376cf09 : SUCCESS in 6m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a87de8ef40d4eb8bbd4d7d5bf36257e : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d418b11c922d4c098021b4b2ff93595f : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/227e7bf89e72406b9f8c8355fc1e7095 : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d83ef733087f4b04bbbf0de7e4763da1 : SUCCESS in 5m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d9e64d35096e427d8b5c01d86c9c352e : RETRY_LIMIT in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f5d8d12dbb04a3e975f01098030617a : SUCCESS in 8m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/027e554168354c5c85759d57fbd180c9 : SUCCESS in 5m 23s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/7fff130a23454d44a3ca3b02b9cfc230 : SUCCESS in 6m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/876e3e29f2454a3582c62c51736b08de : SUCCESS in 6m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fda17557a26844f4ab7829cb212a3b5e : SUCCESS in 6m 17s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9faa55cd7aab4b16a7bcdd1ce8027b8e : SUCCESS in 18m 27s","accounts_in_message":[],"_revision_number":9},{"id":"c3b380588b07420ed01cf9e5ff4c8c77aaf45e80","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-04-14 09:41:49.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"a6f9d0755d9e7f3258e11a8aadd9da88191de4b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 09:42:47.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"874c1c4ca7a025ca649b6dd1efe53e28a1f613a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 10:32:55.000000000","message":"Patch Set 9: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"03c25372bf98349d7c8059bce57f5d57104c03fd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-04-14 13:17:11.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e2abbbc31e94a7d0b55fffe5ffcee58fcafa4ef9","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-04-14 13:49:02.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"539b6b0dab0d28af28e159d350d7b218f7071d78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 14:14:40.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d9cc408e932b4e1eba8d62cd6b340b8c : SUCCESS in 52m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e980204715384e61a7cac9b4d381e66e : SUCCESS in 6m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7c91a3dc30f4eafb0930d517cd569d9 : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf1f02245a68471caed618a7b0fd46ff : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea1d3ed5949a48288d18e8cd208893fa : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72dac9662da54c509e9016aaf8ed05b2 : SUCCESS in 5m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7179a01ecda347059c6808c810c7fdef : RETRY_LIMIT in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d23782f17eaa401d81be2eda0cb249ea : SUCCESS in 7m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0b88d122f66e4925a9a9661a256ef00f : SUCCESS in 6m 08s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/a765c1e528cb4de59e86de9efc2b3cc1 : SUCCESS in 6m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f35e5c322b534f65b4d8ac9303b97906 : SUCCESS in 7m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1cbe864c42ba41c0aa5dc39016b867cd : SUCCESS in 5m 57s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a9f4736c4cbc4f988724e4338cc5c746 : SUCCESS in 16m 46s","accounts_in_message":[],"_revision_number":9},{"id":"f41438651ab0032d36e9397d271713b8d161b07d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 14:32:02.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"4f2cf890a04d543c470bca1ee51fe4170857d357","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 15:56:29.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f4dd2915699402fa585a9445c5b83f1 : SUCCESS in 47m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22ff266b34a641e68a989e81e577a01a : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4fbff967c1c46459f14ae6e9ea969d7 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eccf25defeb4454cabf5eb3a54275f3c : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6365ec45cda54ffbae9f4817e224fbcc : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20a8ee21df854ba9a78c24a5133952c7 : SUCCESS in 8m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d63294af341a4c58b9a1c9ff951a9436 : SUCCESS in 5m 22s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/edd76d5c3ee74e14b4bab62108cce8cc : SUCCESS in 5m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d16fd1bfca0a406db546f4a364f9275a : SUCCESS in 6m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0526f40a2e4c4975a70c0cfd85aec402 : SUCCESS in 6m 27s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/39f135b2672e492ca3c3b87cb385e132 : SUCCESS in 16m 28s","accounts_in_message":[],"_revision_number":9},{"id":"c4891e11927671692d4cc83837053d19edf8639a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 15:56:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"3aca87e4fc305d49db314ce6467f7a63c4e8a526","revisions":{"e5bca904bd5e6ea723f7a7f21913dc84c56636a5":{"kind":"REWORK","_number":1,"created":"2021-07-16 16:50:39.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/05/801105/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/05/801105/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/801105/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/801105/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/801105/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/05/801105/1"}}},"commit":{"parents":[{"commit":"652e9e3c257f7d71ecd52da48aff94e845bb3d99","subject":"db: Remove use of non-integer/slice indices","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/652e9e3c257f7d71ecd52da48aff94e845bb3d99"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-15 12:47:01.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 16:35:48.000000000","tz":60},"subject":"db: Replace deprecated \u0027FromClause.select().whereclause\u0027 parameter","message":"db: Replace deprecated \u0027FromClause.select().whereclause\u0027 parameter\n\nResolve the following sqlalchemy.exc.RemovedIn20Warning warning:\n\n  The FromClause.select().whereclause parameter is deprecated and will\n  be removed in version 2.0.  Please make use of the Select.where()\n  method to add WHERE criteria to the SELECT statement.\n\nFor more information, refer to http://sqlalche.me/e/b8d9.\n\nChange-Id: Ie93f52777da62fa98e8b33d4d1ca0ae6da4de9e0\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e5bca904bd5e6ea723f7a7f21913dc84c56636a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e5bca904bd5e6ea723f7a7f21913dc84c56636a5"}]},"branch":"refs/heads/master"},"7e8a913633f4da41ea3329a2fbbbb8ef93cc9be3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-07-27 17:28:35.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/05/801105/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/05/801105/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/801105/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/801105/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/801105/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/05/801105/2"}}},"commit":{"parents":[{"commit":"ebf5f8d59c829a6f92e978988a3ee5ef47d376be","subject":"db: Remove use of non-integer/slice indices","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ebf5f8d59c829a6f92e978988a3ee5ef47d376be"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-15 12:47:01.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-27 17:19:41.000000000","tz":60},"subject":"db: Replace deprecated \u0027FromClause.select().whereclause\u0027 parameter","message":"db: Replace deprecated \u0027FromClause.select().whereclause\u0027 parameter\n\nResolve the following sqlalchemy.exc.RemovedIn20Warning warning:\n\n  The FromClause.select().whereclause parameter is deprecated and will\n  be removed in version 2.0.  Please make use of the Select.where()\n  method to add WHERE criteria to the SELECT statement.\n\nFor more information, refer to http://sqlalche.me/e/b8d9.\n\nChange-Id: Ie93f52777da62fa98e8b33d4d1ca0ae6da4de9e0\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/7e8a913633f4da41ea3329a2fbbbb8ef93cc9be3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/7e8a913633f4da41ea3329a2fbbbb8ef93cc9be3"}]},"branch":"refs/heads/master"},"db70416aaaa2cd1654665de6365e29a0b5487765":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-07-28 09:44:06.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/05/801105/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/05/801105/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/801105/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/801105/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/801105/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/05/801105/3"}}},"commit":{"parents":[{"commit":"a50ccf5d2e998b6db3c6b4f3dcc4080bb9cfbbca","subject":"db: Remove use of non-integer/slice indices","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/a50ccf5d2e998b6db3c6b4f3dcc4080bb9cfbbca"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-15 12:47:01.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-28 09:43:57.000000000","tz":60},"subject":"db: Replace deprecated \u0027FromClause.select().whereclause\u0027 parameter","message":"db: Replace deprecated \u0027FromClause.select().whereclause\u0027 parameter\n\nResolve the following sqlalchemy.exc.RemovedIn20Warning warning:\n\n  The FromClause.select().whereclause parameter is deprecated and will\n  be removed in version 2.0.  Please make use of the Select.where()\n  method to add WHERE criteria to the SELECT statement.\n\nFor more information, refer to http://sqlalche.me/e/b8d9.\n\nChange-Id: Ie93f52777da62fa98e8b33d4d1ca0ae6da4de9e0\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/db70416aaaa2cd1654665de6365e29a0b5487765"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/db70416aaaa2cd1654665de6365e29a0b5487765"}]},"branch":"refs/heads/master"},"d2433166ed0c5508514df0aaf2bae3516c76d01b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-02-05 16:57:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/801105/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/05/801105/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/801105/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/801105/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/801105/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/05/801105/4"}}},"commit":{"parents":[{"commit":"38fac3a837b83aa212ec9a32b5fb4b7ba69da87c","subject":"db: Remove use of non-integer/slice indices","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/38fac3a837b83aa212ec9a32b5fb4b7ba69da87c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-15 12:47:01.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-05 16:57:36.000000000","tz":0},"subject":"db: Replace deprecated \u0027FromClause.select().whereclause\u0027 parameter","message":"db: Replace deprecated \u0027FromClause.select().whereclause\u0027 parameter\n\nResolve the following sqlalchemy.exc.RemovedIn20Warning warning:\n\n  The FromClause.select().whereclause parameter is deprecated and will\n  be removed in version 2.0.  Please make use of the Select.where()\n  method to add WHERE criteria to the SELECT statement.\n\nFor more information, refer to http://sqlalche.me/e/b8d9.\n\nChange-Id: Ie93f52777da62fa98e8b33d4d1ca0ae6da4de9e0\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d2433166ed0c5508514df0aaf2bae3516c76d01b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d2433166ed0c5508514df0aaf2bae3516c76d01b"}]},"branch":"refs/heads/master","description":"Rebase"},"b773189a2c583db2c3a0a2ed224578c432dab819":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-02-05 17:00:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/801105/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/05/801105/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/801105/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/801105/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/801105/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/05/801105/5"}}},"commit":{"parents":[{"commit":"702498ac3c440943c0807474f95526c175a59176","subject":"db: Remove use of non-integer/slice indices","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/702498ac3c440943c0807474f95526c175a59176"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-15 12:47:01.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-05 17:00:06.000000000","tz":540},"subject":"db: Replace deprecated \u0027FromClause.select().whereclause\u0027 parameter","message":"db: Replace deprecated \u0027FromClause.select().whereclause\u0027 parameter\n\nResolve the following sqlalchemy.exc.RemovedIn20Warning warning:\n\n  The FromClause.select().whereclause parameter is deprecated and will\n  be removed in version 2.0.  Please make use of the Select.where()\n  method to add WHERE criteria to the SELECT statement.\n\nFor more information, refer to http://sqlalche.me/e/b8d9.\n\nChange-Id: Ie93f52777da62fa98e8b33d4d1ca0ae6da4de9e0\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b773189a2c583db2c3a0a2ed224578c432dab819"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b773189a2c583db2c3a0a2ed224578c432dab819"}]},"branch":"refs/heads/master"},"aa50374728859b09d50968ac06bf4b18238b0dc4":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-02-07 12:29:58.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/05/801105/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/05/801105/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/801105/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/801105/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/801105/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/05/801105/6"}}},"commit":{"parents":[{"commit":"8e563e506fd143f54536dec235afa59ef460f59b","subject":"db: Remove use of non-integer/slice indices","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/8e563e506fd143f54536dec235afa59ef460f59b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-15 12:47:01.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-02-07 12:21:28.000000000","tz":0},"subject":"db: Replace deprecated \u0027FromClause.select().whereclause\u0027 parameter","message":"db: Replace deprecated \u0027FromClause.select().whereclause\u0027 parameter\n\nResolve the following sqlalchemy.exc.RemovedIn20Warning warning:\n\n  The FromClause.select().whereclause parameter is deprecated and will\n  be removed in version 2.0.  Please make use of the Select.where()\n  method to add WHERE criteria to the SELECT statement.\n\nFor more information, refer to http://sqlalche.me/e/b8d9.\n\nChange-Id: Ie93f52777da62fa98e8b33d4d1ca0ae6da4de9e0\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/aa50374728859b09d50968ac06bf4b18238b0dc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/aa50374728859b09d50968ac06bf4b18238b0dc4"}]},"branch":"refs/heads/master"},"c6a172bb2eb9b95b676d87f6603a8151bb8882b2":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2022-03-24 15:54:52.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/05/801105/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/05/801105/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/801105/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/801105/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/801105/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/05/801105/7"}}},"commit":{"parents":[{"commit":"40350284b55658aa3e25a1a8e6fb4858203e2600","subject":"db: Remove use of non-integer/slice indices","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/40350284b55658aa3e25a1a8e6fb4858203e2600"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-15 12:47:01.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-03-24 15:44:21.000000000","tz":0},"subject":"db: Replace deprecated \u0027FromClause.select().whereclause\u0027 parameter","message":"db: Replace deprecated \u0027FromClause.select().whereclause\u0027 parameter\n\nResolve the following sqlalchemy.exc.RemovedIn20Warning warning:\n\n  The FromClause.select().whereclause parameter is deprecated and will\n  be removed in version 2.0.  Please make use of the Select.where()\n  method to add WHERE criteria to the SELECT statement.\n\nFor more information, refer to http://sqlalche.me/e/b8d9.\n\nChange-Id: Ie93f52777da62fa98e8b33d4d1ca0ae6da4de9e0\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/c6a172bb2eb9b95b676d87f6603a8151bb8882b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/c6a172bb2eb9b95b676d87f6603a8151bb8882b2"}]},"branch":"refs/heads/master"},"c17b04df9a891497d4eeefa63f05710ab3bb012e":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-03-24 20:14:12.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/05/801105/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/05/801105/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/801105/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/801105/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/801105/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/05/801105/8"}}},"commit":{"parents":[{"commit":"021cac667fd716ae3125fd79b710a84730c488f7","subject":"db: Remove use of non-integer/slice indices","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/021cac667fd716ae3125fd79b710a84730c488f7"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-15 12:47:01.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-03-24 20:14:04.000000000","tz":0},"subject":"db: Replace deprecated \u0027FromClause.select().whereclause\u0027 parameter","message":"db: Replace deprecated \u0027FromClause.select().whereclause\u0027 parameter\n\nResolve the following sqlalchemy.exc.RemovedIn20Warning warning:\n\n  The FromClause.select().whereclause parameter is deprecated and will\n  be removed in version 2.0.  Please make use of the Select.where()\n  method to add WHERE criteria to the SELECT statement.\n\nFor more information, refer to http://sqlalche.me/e/b8d9.\n\nChange-Id: Ie93f52777da62fa98e8b33d4d1ca0ae6da4de9e0\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/c17b04df9a891497d4eeefa63f05710ab3bb012e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/c17b04df9a891497d4eeefa63f05710ab3bb012e"}]},"branch":"refs/heads/master"},"3aca87e4fc305d49db314ce6467f7a63c4e8a526":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2022-04-07 14:40:56.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/05/801105/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/05/801105/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/801105/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/801105/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/801105/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/05/801105/9"}}},"commit":{"parents":[{"commit":"b3fe04f081a096258468d032560f46cdfe77e144","subject":"db: Remove use of non-integer/slice indices","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b3fe04f081a096258468d032560f46cdfe77e144"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-15 12:47:01.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-04-07 14:39:27.000000000","tz":60},"subject":"db: Replace deprecated \u0027FromClause.select().whereclause\u0027 parameter","message":"db: Replace deprecated \u0027FromClause.select().whereclause\u0027 parameter\n\nResolve the following sqlalchemy.exc.RemovedIn20Warning warning:\n\n  The FromClause.select().whereclause parameter is deprecated and will\n  be removed in version 2.0.  Please make use of the Select.where()\n  method to add WHERE criteria to the SELECT statement.\n\nFor more information, refer to http://sqlalche.me/e/b8d9.\n\nChange-Id: Ie93f52777da62fa98e8b33d4d1ca0ae6da4de9e0\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/3aca87e4fc305d49db314ce6467f7a63c4e8a526"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/3aca87e4fc305d49db314ce6467f7a63c4e8a526"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
