)]}'
{"id":"openstack%2Fplacement~806730","triplet_id":"openstack%2Fplacement~master~I80df62dae3a31481c71254e1054973df39bec955","project":"openstack/placement","branch":"master","topic":"story-2009159","hashtags":[],"change_id":"I80df62dae3a31481c71254e1054973df39bec955","subject":"Add reproducer for Project creation race bug","status":"MERGED","created":"2021-08-31 11:46:50.000000000","updated":"2021-09-03 16:28:07.000000000","submitted":"2021-09-03 16:13:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":8,"has_review_started":true,"submission_id":"806730-story-2009159","meta_rev_id":"31c257d5aa72c30e71c6c80b6ab84dbb0e8861aa","_number":806730,"virtual_id_number":806730,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-03 16:13:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-03 07:49:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2021-09-03 14:54:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2021-09-03 14:54:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-31 12:02:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-01 16:26:52.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2021-09-02 09:23:09.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"}],"messages":[{"id":"1f09ae001472cd2e1e79e8e06f649118d2de85bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-08-31 11:46:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4d69a46c6a71a18547927ea3bde2f52739284fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 12:02:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2eb6f74fc4b049e2a2fcc69f84c7862f : SUCCESS in 7m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb969ce873a04b689f7cf0d136474505 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39c891c7765b4dc3b654e202e54b6037 : SUCCESS in 8m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be3be7e8509f47d7be1a09dbe2800b1f : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4f97341da324cc6a301c8e387a0e209 : SUCCESS in 9m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2226ca5ebf5e4d4387c4ef4fec1eb754 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d396f4779a95442bbfe8ef7608f183a4 : SUCCESS in 10m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/541d29d960064aa180cbbd4daee61ba9 : SUCCESS in 7m 16s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/cb4d831bb9c4455e864d2caae94ca053 : SUCCESS in 9m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/edc6c04ad0b746868f3b8538dfc1758c : SUCCESS in 7m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/33a9d12fea304b0b86c25f66727aeb3c : SUCCESS in 7m 46s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/63b02cc0d6e54cc0a66783c8bddb846d : SUCCESS in 14m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c2fd0b4d57cdb3deeec182c4d93038b1b4ca198d","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-09-01 16:26:52.000000000","message":"Patch Set 1: Code-Review+2\n\nAwesome test, LGTM","accounts_in_message":[],"_revision_number":1},{"id":"0825ead443d19711ae940326310e11b38b44af75","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-09-01 17:32:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4cad89d11f17be6cdcbbd202b9deb3a1bab9167f","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-09-01 17:33:14.000000000","message":"Patch Set 1: -Code-Review\n\nGoing to hold for now while I work on a fix (in case the fix refactor will affect the test).","accounts_in_message":[],"_revision_number":1},{"id":"968cb11d3fdc359a92766a5378be8362a26e0433","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-09-01 18:44:33.000000000","message":"Patch Set 1:\n\nAlso note that this patch\u0027s parent change is from before the followup patch merged:\n\nhttps://review.opendev.org/c/openstack/placement/+/806035\n\nand it involved changes to the code that the fix will need to change. For that reason, I will rebase this patch at the same time I upload a fix patch.","accounts_in_message":[],"_revision_number":1},{"id":"035143923aa77cd854c920efad56ca5dd206cb55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-09-02 02:50:17.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"2fd8f77c07c1ea71699c755a663b798fdf7df1e3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-09-02 02:51:03.000000000","message":"Topic story-2009159-rev3 removed","accounts_in_message":[],"_revision_number":2},{"id":"5a1f7d7e025ada5ca845c3df9e8ba82d5525a0dd","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-09-02 02:51:10.000000000","message":"Topic set to story-2009159","accounts_in_message":[],"_revision_number":2},{"id":"80d204a7491e3018ea0566782c21f76ad06b6fcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 03:30:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ddb594b2ecd645f291c6dfa516125b4d : SUCCESS in 18m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/503b882e131847168a0508f3155ad39a : SUCCESS in 15m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0828e4edb6f3426fa64054f1f52d27ea : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6ec60affd4a4f9d80b079752475e0d2 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1feb6a3d7b8e478cabc4ae1a2fcbd700 : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4b0804a133d46b1b45f2744a59e25d9 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0fc5b05a7de4b3a95d3c1f8bade9bd7 : SUCCESS in 7m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/21c12b4cacec4ee1a837a6e849bb867c : SUCCESS in 5m 56s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/f5e4aecd42b14594af0e9dcceed14621 : SUCCESS in 6m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/01dcaf0927e74422b99fd021c4916434 : SUCCESS in 6m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cfb4da32b2d943dc98d730c3606abd9c : SUCCESS in 7m 07s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/560023ddc1d84c44a2607a5e29ba0bbf : SUCCESS in 14m 48s","accounts_in_message":[],"_revision_number":2},{"id":"39fd9c8364c690055ab56b7a7e82ca7a51b4027c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-09-02 08:45:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"07b3e6f44acb0cbe86c1c73c238ff3960c2b28e2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-09-02 08:45:42.000000000","message":"Topic story-2009159-rev3 removed","accounts_in_message":[],"_revision_number":3},{"id":"5102b275f91923f8d8cb7f029b6ac2c80aa82e54","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-09-02 08:45:46.000000000","message":"Topic set to story-2009159","accounts_in_message":[],"_revision_number":3},{"id":"90a5986ebda3d24fe1fc642a87dcb34fc2f81314","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 09:03:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d20d1c5aafcb4294b6e6050f065d5abe : SUCCESS in 7m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab77f815eca64936b649cb54d0f53f5b : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1450c5c17d74ce9b5354c4027c79ae1 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7920e4a220e647d9862f37903ce01eb7 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96f65096f18d446f91e5830ace089f08 : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c752e73b3ba04f8789ce60abb64f32ff : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aea96418121c49b5b9ed2938b69d1d38 : SUCCESS in 9m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c9b1db4a82434426b556ca0ef7a369db : SUCCESS in 6m 48s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/91c821952f3e4a3b89b913412a1e0e52 : SUCCESS in 5m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e9857ce5f1dd4b20b97f7f1438997f41 : SUCCESS in 7m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b3aabc9268284c9088a779f59c0cab2a : SUCCESS in 6m 40s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/243719b287db45b2aa960d6b5aa33abc : SUCCESS in 14m 53s","accounts_in_message":[],"_revision_number":3},{"id":"200052315413bab8c1e5cf962d8741c568316205","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2021-09-02 09:23:09.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)\n\nThis looks a bit hacky, but OK.","accounts_in_message":[],"_revision_number":3},{"id":"72848ef6778b5ee83e2510b76c1ec3b8fcd7ac02","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-09-02 09:43:11.000000000","message":"Patch Set 3: Code-Review+1\n\nThe change since my patch looks OK.","accounts_in_message":[],"_revision_number":3},{"id":"93b2b56ad67cdab74cf4781ed7662ae575012b2e","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-09-02 17:32:06.000000000","message":"Patch Set 3: Workflow-1\n\nGoing to try something different bc I think I did a lot of wrong patterns in this.","accounts_in_message":[],"_revision_number":3},{"id":"011cc55b12aeabfd6cc2eb1df244fa171323dd13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-09-02 20:18:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"59e1fc3873aa0c07d0bff478e481c9cfb1e41788","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 20:38:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aea0de88cdd94011b12ecf802b6874e0 : SUCCESS in 9m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c7d0a40847e48b596f17f9b6c7a4bb7 : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5303bee29c2549e2b60e7340474e74c4 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/781ecd2353eb464f854a1d0c5903a0ae : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e339cd0d7fad4719b4b0b06aff14d093 : SUCCESS in 7m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd65d2d1cffd4b0f960bf3194fae9002 : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d93751729aa4cc5b0aa66cbbdb946fb : SUCCESS in 9m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/719fa831d3e941168e3b70f837a70b55 : SUCCESS in 7m 03s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/c39cc753687e4945b77ccce3f0a20e1b : SUCCESS in 7m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6e9a1dbc96924a7bbb4168bbb790d8d8 : SUCCESS in 6m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ada221ec732143429be237af028999c1 : SUCCESS in 6m 19s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1f42b1d459f949b4bebe00b12278bfdb : SUCCESS in 13m 55s","accounts_in_message":[],"_revision_number":4},{"id":"5b3f5ef2d9707ba1fb5461de498a67ab8f8babb3","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-09-03 07:49:46.000000000","message":"Patch Set 4: Code-Review+2\n\nI agree, this is cleaner. Thanks.","accounts_in_message":[],"_revision_number":4},{"id":"f45a1c0424ed883e9437b61e4e76112ff6764291","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2021-09-03 14:54:05.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":4},{"id":"d48354089adcb33b8874eac32a72cf7db7849229","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 14:54:42.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ae56f157120786226c4635e5a374addf0a66b43a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 16:13:48.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27aa0a4138c54267b1716feace1ba73b : SUCCESS in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99a245c423ff44ddb8525dd53c229062 : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c36ef9ea3974070bd9dc83994cdfdce : SUCCESS in 8m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09b040aa4be24fb2a4712634820999e0 : SUCCESS in 9m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65b061dd04b24e8497ce4a61792945a4 : SUCCESS in 10m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/12ccf1867b2c44a49c3e28c5636dac1b : SUCCESS in 7m 06s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/124c78ecc8db454a9b3e7e7d0957050c : SUCCESS in 7m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cf69beca1d9a4a8994242f7dfefcda5d : SUCCESS in 8m 16s\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ea2e2416a16d4f45938916a3145c3812 : SUCCESS in 17m 22s","accounts_in_message":[],"_revision_number":4},{"id":"9a6b848e1a8d10a86c2f7476db98c7922be8e9cf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 16:13:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"ed2b4710100504309f027e166c92daead9aeca0e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 16:16:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8b64c2def7e41f2b6aa57923c7ff2ca : SUCCESS in 1m 53s","accounts_in_message":[],"_revision_number":4},{"id":"31c257d5aa72c30e71c6c80b6ab84dbb0e8861aa","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-09-03 16:28:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8b000867ca50ca99877fe8819785918aa6f58764","revisions":{"039b36024dd5b7e2f7e913119c4476e6d3c4c861":{"kind":"REWORK","_number":1,"created":"2021-08-31 11:46:50.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/30/806730/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/30/806730/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/30/806730/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/30/806730/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/30/806730/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/30/806730/1"}}},"commit":{"parents":[{"commit":"48f31d446be5dd8743392e6d1e45ed8183a9ce1b","subject":"Merge \"Switch ConsumerType to use an AttributeCache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/48f31d446be5dd8743392e6d1e45ed8183a9ce1b"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-08-31 11:39:38.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-08-31 11:46:09.000000000","tz":120},"subject":"Add reproducer for Project creation race bug","message":"Add reproducer for Project creation race bug\n\nThis patch adds a functional tests that reproduce the error\n\n  sqlalchemy.exc.InvalidRequestError: This session is in \u0027inactive\u0027\n  state, due to the SQL transaction being rolled back; no further SQL\n  can be emitted within this transaction.\n\nthat happens during two racing allocation update.\n\nStory: 2009159\nTask: 43147\nChange-Id: I80df62dae3a31481c71254e1054973df39bec955\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/039b36024dd5b7e2f7e913119c4476e6d3c4c861"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/039b36024dd5b7e2f7e913119c4476e6d3c4c861"}]},"branch":"refs/heads/master"},"8d46b50dc56ee963de0f49542828643e42af7297":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-09-02 02:50:17.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/30/806730/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/30/806730/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/30/806730/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/30/806730/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/30/806730/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/30/806730/2"}}},"commit":{"parents":[{"commit":"8db5d027c361fae644bfe3b3b9c8cfe8285efa66","subject":"Merge \"Refactor consumer type methods for readability\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/8db5d027c361fae644bfe3b3b9c8cfe8285efa66"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-08-31 11:39:38.000000000","tz":120},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2021-09-01 22:05:01.000000000","tz":0},"subject":"Add reproducer for Project creation race bug","message":"Add reproducer for Project creation race bug\n\nThis patch adds a functional tests that reproduce the error\n\n  sqlalchemy.exc.InvalidRequestError: This session is in \u0027inactive\u0027\n  state, due to the SQL transaction being rolled back; no further SQL\n  can be emitted within this transaction.\n\nthat happens during two racing allocation update.\n\nStory: 2009159\nTask: 43147\nChange-Id: I80df62dae3a31481c71254e1054973df39bec955\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/8d46b50dc56ee963de0f49542828643e42af7297"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/8d46b50dc56ee963de0f49542828643e42af7297"}]},"branch":"refs/heads/master"},"2786060fac6aaa9db26409305ed357e7800396cc":{"kind":"REWORK","_number":3,"created":"2021-09-02 08:45:01.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/30/806730/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/30/806730/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/30/806730/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/30/806730/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/30/806730/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/30/806730/3"}}},"commit":{"parents":[{"commit":"8db5d027c361fae644bfe3b3b9c8cfe8285efa66","subject":"Merge \"Refactor consumer type methods for readability\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/8db5d027c361fae644bfe3b3b9c8cfe8285efa66"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-08-31 11:39:38.000000000","tz":120},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2021-09-02 08:33:18.000000000","tz":0},"subject":"Add reproducer for Project creation race bug","message":"Add reproducer for Project creation race bug\n\nThis patch adds a functional tests that reproduce the error\n\n  sqlalchemy.exc.InvalidRequestError: This session is in \u0027inactive\u0027\n  state, due to the SQL transaction being rolled back; no further SQL\n  can be emitted within this transaction.\n\nthat happens during two racing allocation update.\n\nStory: 2009159\nTask: 43147\nChange-Id: I80df62dae3a31481c71254e1054973df39bec955\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/2786060fac6aaa9db26409305ed357e7800396cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/2786060fac6aaa9db26409305ed357e7800396cc"}]},"branch":"refs/heads/master"},"8b000867ca50ca99877fe8819785918aa6f58764":{"kind":"REWORK","_number":4,"created":"2021-09-02 20:18:12.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/30/806730/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/30/806730/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/30/806730/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/30/806730/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/30/806730/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/30/806730/4"}}},"commit":{"parents":[{"commit":"8db5d027c361fae644bfe3b3b9c8cfe8285efa66","subject":"Merge \"Refactor consumer type methods for readability\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/8db5d027c361fae644bfe3b3b9c8cfe8285efa66"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-08-31 11:39:38.000000000","tz":120},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2021-09-02 20:14:56.000000000","tz":0},"subject":"Add reproducer for Project creation race bug","message":"Add reproducer for Project creation race bug\n\nThis patch adds a functional tests that reproduce the error\n\n  sqlalchemy.exc.InvalidRequestError: This session is in \u0027inactive\u0027\n  state, due to the SQL transaction being rolled back; no further SQL\n  can be emitted within this transaction.\n\nthat happens during two racing allocation update.\n\nCo-Authored-By: melanie witt \u003cmelwittt@gmail.com\u003e\n\nStory: 2009159\nTask: 43147\n\nChange-Id: I80df62dae3a31481c71254e1054973df39bec955\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/8b000867ca50ca99877fe8819785918aa6f58764"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/8b000867ca50ca99877fe8819785918aa6f58764"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
