)]}'
{"id":"openstack%2Fplacement~809994","triplet_id":"openstack%2Fplacement~master~If79c44656c1a024e0e433da198daea96f64b296c","project":"openstack/placement","branch":"master","hashtags":[],"change_id":"If79c44656c1a024e0e433da198daea96f64b296c","subject":"[DNM]: Trigger lower-constaints job","status":"ABANDONED","created":"2021-09-20 10:07:51.000000000","updated":"2021-09-20 12:17:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"df42966824d220c775b06f34b15a618520e52763","_number":809994,"virtual_id_number":809994,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-20 11:13:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-20 11:13:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d84c98b4aa2bc02b2042404d7bcca7944cd0bec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-09-20 10:07:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15ce4cbd8f59a4e7ca8b64ab29144d5982341e02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 11:13:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3e564389e72d41aab65356eda4cdee4c : SUCCESS in 7m 45s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6ff8e19ceadc460fa544068cabc7ec86 : SUCCESS in 2m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b3da699c93104c779bf047d8ed9f3e24 : SUCCESS in 6m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/cc3f41cdae544539b877479931ac9a14 : SUCCESS in 58m 56s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/31057926b92e423897c099022183fafb : SUCCESS in 1h 03m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9bf9421babd4d57b75642abd270981b : SUCCESS in 6m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c8f98841ae14f1db8395cf4ca1ebb21 : TIMED_OUT in 41m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88385b0959e1422580149f01902f61ce : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2698893505cf4c4ea63b3b1e4196abdd : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/efc026dcff434ceda11f92bb67c5e456 : SUCCESS in 8m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0ca2b11d68f456e95f27e67f4a3d418 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f6f8c2df7af4b26a83807dc788e46f8 : SUCCESS in 9m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d875601c33b94a6aae8e80884161f8d8 : SUCCESS in 6m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/60ec717efb36458dbc96582952d93e98 : SUCCESS in 7m 09s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/3cbd76488b1c45028b8e6a3a21753c37 : SUCCESS in 7m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/275f39b2227f4a61908e4716e9668890 : SUCCESS in 6m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1b9699d896a840aaaf87be191b0e5421 : SUCCESS in 7m 04s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8473c9727c554c8ea5594a4fdca593c5 : SUCCESS in 14m 46s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/4c5137ebf46845b7a10c5150c2db729a : SUCCESS in 23m 14s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/be6c641eb0ef47a5aa3c0a00e0fdd888 : SUCCESS in 9m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/187ec77c0d0e49c993f03a46a5a211d0 : SUCCESS in 49m 39s","accounts_in_message":[],"_revision_number":1},{"id":"df42966824d220c775b06f34b15a618520e52763","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-09-20 12:17:06.000000000","message":"Abandoned\n\nyepp lower is broken fix proposed in https://review.opendev.org/c/openstack/placement/+/810001","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"02ba214c449f2fa46f27da8a4c3aa5683ce56c5d","revisions":{"02ba214c449f2fa46f27da8a4c3aa5683ce56c5d":{"kind":"REWORK","_number":1,"created":"2021-09-20 10:07:51.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/94/809994/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/94/809994/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/94/809994/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/94/809994/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/94/809994/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/94/809994/1"}}},"commit":{"parents":[{"commit":"de4eb283d391f3e94b1a60edc7fd72e1ebba76ac","subject":"Update master for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/de4eb283d391f3e94b1a60edc7fd72e1ebba76ac"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-09-20 10:06:50.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-09-20 10:06:50.000000000","tz":120},"subject":"[DNM]: Trigger lower-constaints job","message":"[DNM]: Trigger lower-constaints job\n\nChange-Id: If79c44656c1a024e0e433da198daea96f64b296c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/02ba214c449f2fa46f27da8a4c3aa5683ce56c5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/02ba214c449f2fa46f27da8a4c3aa5683ce56c5d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
