)]}'
{"id":"openstack%2Fplacement~819206","triplet_id":"openstack%2Fplacement~master~I50c125df4ee03377caf16da5824e2ed492d4b7da","project":"openstack/placement","branch":"master","topic":"yoga-python-testing","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2021-12-20 19:32:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I50c125df4ee03377caf16da5824e2ed492d4b7da","subject":"Updating python testing as per Yoga testing runtime","status":"MERGED","created":"2021-11-25 00:08:15.000000000","updated":"2021-12-20 19:47:35.000000000","submitted":"2021-12-20 19:32:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"819206-yoga-python-testing","meta_rev_id":"8e84296f591c34d089562e5f73060a39c27d418c","_number":819206,"virtual_id_number":819206,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-20 19:32:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"recommended":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"all":[{"value":1,"date":"2021-12-18 16:05:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":2,"date":"2021-12-14 12:19:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2021-12-20 15:50:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2021-12-20 15:50:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":1,"date":"2021-12-20 19:47:35.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2021-12-14 16:16:10.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-25 01:12:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-26 08:48:42.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-12-14 16:16:10.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2021-12-18 16:05:25.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"}],"messages":[{"id":"944ee5f4580a0a2e7232319d6c809c0383f4ae58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-25 00:08:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ad4852c3c00c903c18e25a1e916273d43e5ad6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 01:12:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cf06abcf47bd42ad9656c7410a7bd159 : SUCCESS in 6m 35s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e9242b9da790461baae429368a75bcd4 : SUCCESS in 2m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/c9509ab4edb1432eaed9da2eaebe6565 : SUCCESS in 52m 37s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/c9e35cb384204df7b63ca94891fa5810 : SUCCESS in 1h 01m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97b07e18a48f49dea108438ef370076a : SUCCESS in 6m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3208c901909d4b268f12696b537cff5b : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b94fb345aa741cd876d3166aef9464e : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cb686a8dd334197ae79ed3def4bef74 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f2262f21261446ea394dc2e68abb0fe : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9733d0e62d544408ac877c8dad3f7a1d : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c125ca7754924860a5fc46e30095407a : SUCCESS in 6m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/13cfa09c54db4074b3373793ce5f3135 : SUCCESS in 7m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c85360fc265540f8a9770bd31b5a65d9 : SUCCESS in 8m 51s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b9fb761fac9942b29b3c992ab0cca8cd : SUCCESS in 15m 47s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/afa66f3fc3c041fba3b92ace0d7c8d07 : SUCCESS in 17m 13s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/2a3cc59639574285bd5baf7b8f3bafb0 : SUCCESS in 10m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd7a7f2a34114c25bc77b99c5bc30fe5 : SUCCESS in 54m 05s","accounts_in_message":[],"_revision_number":1},{"id":"381ee04d3855d03ff36b7e4469e29d8105b26b05","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-11-26 08:48:42.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"ef33a2e1120e72ea4f0819fa8d49fce060f57c0d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-26 16:02:35.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b2a0e66c0b63b66dc7b5870827bc1ada4a266bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-12-14 02:24:04.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"582e545ce6ef3235629704082f543bcd523458f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 04:24:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d77a1845d2144af78b43a5d370c6f6c6 : SUCCESS in 8m 55s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a27dd8c15dc5406eaabb49f21e949075 : SUCCESS in 3m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/24952b22fadc4c5fb6e3e7bae50c5be7 : SUCCESS in 58m 26s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/c4015e6f772c48308b13573cc33301da : SUCCESS in 1h 14m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2d5866cb0ae444a9d1bb717e24e6bc0 : SUCCESS in 49m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbf821cb198b41af86221539ce51d504 : SUCCESS in 9m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41b0c49454d14c5297c1f4f81d23e45f : SUCCESS in 8m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/959d7e3575a44df09f77a8b17fe0bbc8 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/717e9313e41e4661ae7cc2f19aafb433 : SUCCESS in 7m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b90ce12defde4cf59e54624b7c4ef527 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/567dca422364488c995408b420c29fb1 : SUCCESS in 13m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b458869aadb40eeb99023ead162030c : SUCCESS in 7m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aca1cda2126140f78ab3b82d717dd04b : SUCCESS in 7m 05s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/cc16e2b042914813bb95f0e4cf105681 : SUCCESS in 8m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/995267e98190417b9f82a7324e488a3f : SUCCESS in 7m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/42d2d1bb5de74db08726af59aba0f20e : SUCCESS in 7m 09s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/aa37f31307334e5fb4da71ed3ae77ecf : SUCCESS in 22m 00s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/c1c82b916e064e54bacb06c301fd3e80 : SUCCESS in 27m 07s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/9c5da9f1a1be4294a34ca9242022db23 : SUCCESS in 10m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98735725406744acaa6d81659991280e : SUCCESS in 55m 58s","accounts_in_message":[],"_revision_number":2},{"id":"7ddff81d502ffcab007b88cb3e1b4d0fea7706b4","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-12-14 12:19:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"706cb7a2ed92c79f61995f27aae1401cdfd982e7","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2021-12-14 16:16:10.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"9265da159e3e60c661eaa8c778ca2d3732e70085","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2021-12-18 16:05:25.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a20b37731680d030872309d5b32b29e7029704a0","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2021-12-20 15:50:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"085bf25cef5a7afcbfe6483ca45b80268c89a262","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 15:52:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7df10ba65969ab524670939a699d4142b10713ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 19:32:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/84abf20dbf1949cd93b0985aeefd7a54 : SUCCESS in 5m 56s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c4be62169a624f7596f2cf7b09de1142 : SUCCESS in 2m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/ea4a2e4bf6564df7b6d70beadd371561 : SUCCESS in 55m 26s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/cd32fb05a89d46968f6c061af1c6ab64 : SUCCESS in 1h 12m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/932403961a23404dbda89766dd3d2067 : SUCCESS in 51m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4eb58e4f268c41f1ab39c02fbd3b8fd6 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68eeb73f7ba84845baf72ca3cca33ad6 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b470875fb53f4648a76f42da8de50354 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30e9deb33d214fc8a87e673fe6b9e79e : SUCCESS in 8m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/854fd465655e4acfa85278c2cdf7dd69 : SUCCESS in 10m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbcd763b970040bda96bc12fbb9fe0ee : SUCCESS in 6m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e456c78924824a179af0492a469dba24 : SUCCESS in 6m 39s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/17fa3ef6bab941e8b80318b2267f6143 : SUCCESS in 7m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/08bfe126b3cf442091e7ea04dcfa1405 : SUCCESS in 7m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5b71aaea573f405aa16d793fd5a5ab13 : SUCCESS in 6m 55s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ca77b08fbd5247f58d15fd122514dfe8 : SUCCESS in 18m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b253c681a292456a95f716fcdcb91e75 : SUCCESS in 47m 48s","accounts_in_message":[],"_revision_number":2},{"id":"aec41c6429ebf3c059dbe77f1c8ab9a2d6b7772f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 19:32:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1b1f8727f8426f83befdfaef8ba77213e9994efb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 19:47:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/89da7d30a1664cef91cf58c50f52154b : SUCCESS in 1m 42s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9989b1c96ccf4df5a6765a7b62c1919e : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf6781d9ec4c480f974f802a6cf1e954 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1141597b5a079ebd2d0304394d24526163be2b42","revisions":{"efc2a2b9b8c449ae37c9c1d78be1d0647861f4fe":{"kind":"REWORK","_number":1,"created":"2021-11-25 00:08:15.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/06/819206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/06/819206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/06/819206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/06/819206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/06/819206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/06/819206/1"}}},"commit":{"parents":[{"commit":"6d49b6df1d3f62e2003c55509875380148281a79","subject":"Use \u0027functional-without-sample-db-tests\u0027 tox env for placement nova job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/6d49b6df1d3f62e2003c55509875380148281a79"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-25 00:02:10.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-25 00:02:10.000000000","tz":-360},"subject":"Updating python testing as per Yoga testing runtime","message":"Updating python testing as per Yoga testing runtime\n\nYoga testing runtime has been updated with py38 and py39\nas voting and removed the py36 testing. Unit tests update are\nhandled by the job template change in openstack-zuul-job\n\n- https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/818609\n\nthis commit makes other required changes in fucntional jobs and\nupdate the metadata in setup.cfg file.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/yoga.html\n\nChange-Id: I50c125df4ee03377caf16da5824e2ed492d4b7da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/efc2a2b9b8c449ae37c9c1d78be1d0647861f4fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/efc2a2b9b8c449ae37c9c1d78be1d0647861f4fe"}]},"branch":"refs/heads/master"},"1141597b5a079ebd2d0304394d24526163be2b42":{"kind":"REWORK","_number":2,"created":"2021-12-14 02:24:04.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/06/819206/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/06/819206/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/06/819206/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/06/819206/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/06/819206/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/06/819206/2"}}},"commit":{"parents":[{"commit":"6d49b6df1d3f62e2003c55509875380148281a79","subject":"Use \u0027functional-without-sample-db-tests\u0027 tox env for placement nova job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/6d49b6df1d3f62e2003c55509875380148281a79"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-25 00:02:10.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2021-12-14 02:24:04.000000000","tz":0},"subject":"Updating python testing as per Yoga testing runtime","message":"Updating python testing as per Yoga testing runtime\n\nYoga testing runtime[1] has been updated to add py39\ntesting as voting. Unit tests update are\nhandled by the job template change in openstack-zuul-job\n\n- https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/820286\n\nthis commit makes other required changes in zuul.yaml and\nupdate the classifier in setup.cfg file.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/yoga.html\n\nChange-Id: I50c125df4ee03377caf16da5824e2ed492d4b7da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/1141597b5a079ebd2d0304394d24526163be2b42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/1141597b5a079ebd2d0304394d24526163be2b42"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
