)]}'
{"id":"openstack%2Fplacement~819766","triplet_id":"openstack%2Fplacement~master~I44f6d46205200d20dba57c9d0d77ab5a56b426b6","project":"openstack/placement","branch":"master","hashtags":[],"change_id":"I44f6d46205200d20dba57c9d0d77ab5a56b426b6","subject":"DNM: check if doc build is broken on master","status":"ABANDONED","created":"2021-11-30 08:51:34.000000000","updated":"2021-11-30 10:17:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2f068a3fa9d927dea53d33ed7fb0b091561578bf","_number":819766,"virtual_id_number":819766,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-11-30 10:09:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-30 10:09:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5bfe0d4600596c3cb93139ab3988ca69131b9e83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-11-30 08:51:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b037034203b767707f053066e8108627c4f55ea6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 10:09:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/41907fd33e0f483aa86b24be0ab37744 : SUCCESS in 1h 14m 50s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/859f2afe61ef4978abf0fd3d6aafd080 : SUCCESS in 1h 02m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a9b5087fb6d4f63bb01e2df4120f859 : SUCCESS in 8m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d019d242042d4acf81907f92e844805a : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/759dc6b1cd334ac9a99278a8e97c270c : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0acdf52bb8a9410394bd758dfb8fd8b8 : SUCCESS in 6m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e36876f236e4f08b2a1d9472432a00a : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5be53c89ca3f43b7a466d14d03243f76 : SUCCESS in 8m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/352da1b9e5cc45e4a745f50720be4a1c : SUCCESS in 5m 31s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/04469be914b740ffbd65beecdf0de1a2 : SUCCESS in 5m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a9fcc18731f24dbf8e645d9744f11523 : SUCCESS in 7m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c22cb23f4ffa466182d1db3c2595f321 : SUCCESS in 6m 54s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d43941c2ca624306a897d7163bd4467a : SUCCESS in 16m 34s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/c98fb54951c2442aa11deb83461ea2eb : SUCCESS in 27m 40s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/694ec8d8905b4eb68586442337b0fb18 : SUCCESS in 9m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80660cd0ecab4b708c770f661d49d1dc : SUCCESS in 50m 49s","accounts_in_message":[],"_revision_number":1},{"id":"2f068a3fa9d927dea53d33ed7fb0b091561578bf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-11-30 10:17:32.000000000","message":"Abandoned\n\nno it is not broken, it seems the the dependency issues are only hitting me locally :/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e98797251eb8c7ac07552981fd68178f1cde4de2","revisions":{"e98797251eb8c7ac07552981fd68178f1cde4de2":{"kind":"REWORK","_number":1,"created":"2021-11-30 08:51:34.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/66/819766/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/66/819766/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/66/819766/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/66/819766/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/66/819766/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/66/819766/1"}}},"commit":{"parents":[{"commit":"6e8c4cbffc05f5696bf3a728a4e73d400185fa20","subject":"Merge \"Enable HTTPProxyToWSGI middleware to find actual client ips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/6e8c4cbffc05f5696bf3a728a4e73d400185fa20"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-11-30 08:37:48.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-11-30 08:51:20.000000000","tz":60},"subject":"DNM: check if doc build is broken on master","message":"DNM: check if doc build is broken on master\n\nChange-Id: I44f6d46205200d20dba57c9d0d77ab5a56b426b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e98797251eb8c7ac07552981fd68178f1cde4de2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e98797251eb8c7ac07552981fd68178f1cde4de2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
