)]}'
{"id":"openstack%2Fplacement~840703","triplet_id":"openstack%2Fplacement~stable%2Fussuri~I569043dd9f1c964faf7d4fcbb901a93f9569bf3d","project":"openstack/placement","branch":"stable/ussuri","topic":"bug/1780799","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2023-01-05 02:18:02.000000000","reason":"\u003cGERRIT_ACCOUNT_4690\u003e replied on the change","reason_account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-01-10 14:04:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I569043dd9f1c964faf7d4fcbb901a93f9569bf3d","subject":"placement-status: check only consumers in allocation table","status":"MERGED","created":"2022-05-05 16:30:17.000000000","updated":"2023-01-10 14:05:41.000000000","submitted":"2023-01-10 14:04:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840703-bug/1780799","meta_rev_id":"9dfbe9dea2b4082eb9e2b94cd460383a39d237a5","_number":840703,"virtual_id_number":840703,"owner":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-10 14:04:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2023-01-10 12:38:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2023-01-10 12:38:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"}],"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-05 16:30:17.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"state":"CC"},{"updated":"2022-05-05 17:36:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-20 13:14:13.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2023-01-10 12:38:36.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"d46ad4e48d58a822e15448ea37581cad08785dc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2022-05-05 16:30:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66703aa6de4ba7a5f0271da8deea2806f53c17cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 17:36:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/465d5a3357724bcc88fe27ae73130e7e : SUCCESS in 1h 01m 17s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8e0e103dab514313acb76cd111585f4a : FAILURE in 15m 26s (non-voting)\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/314e4280c4564dff9feae83209a00af7 : SUCCESS in 1h 04m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00cde164a35e48a58fb29f68684de804 : SUCCESS in 47m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36561dcf08cc47f486a0dd220f7834e7 : SUCCESS in 5m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a25af4f17b9c4f19859a5e64481212e3 : RETRY_LIMIT in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cd2ce77600b42fe8897614b7ef4f7ee : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6777499e920b4c6b9ed89aaf859be56f : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7f58bd19064b40c49989efca15ea8a08 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/244b7d099d514fee8e0eb84f540bc4a3 : SUCCESS in 7m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a1b74a3309f24ef4abbca08fabed4db5 : SUCCESS in 5m 14s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/402a6dbc221141519e9633cfd575ed34 : SUCCESS in 6m 30s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/f6f5f72863484c1da56399e78e0ce7d9 : FAILURE in 6m 26s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/1c3d9a28997b4b3cba7d7140a92885a8 : FAILURE in 2m 17s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/0aa0ffb3db944baeb474c339fa4c8cb4 : FAILURE in 2m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9abfbbfc86104edab7b996e60f402010 : SUCCESS in 49m 59s","accounts_in_message":[],"_revision_number":1},{"id":"8ea719f37eee8c9db7ffd7b0bfdc1d613bbbca8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2022-05-05 22:10:58.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"db27068f6e1203a23f77ec3f94057e8b249c2a4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 23:18:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/6730efb3817a45039cd84aba0474313f : SUCCESS in 53m 59s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5e24f333482e4c43ab15b1f420cd8a6d : FAILURE in 13m 02s (non-voting)\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/d2323f03354e486aaa7c74eb0f1c2a8e : SUCCESS in 59m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3c8fcad97ce4ff2933aea57535fc988 : SUCCESS in 44m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d46870883b94a05a1cad2d3ca9df1f0 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad4478af1a834187aee9d8c25e94bbe0 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e734fb080dd749b3a02ce575f12bb0d2 : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cd3db069a34244aabf1f1f3432d1d04b : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ade26e133f654fbbb660ade185d16ec3 : SUCCESS in 7m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b799242af39a47b196869f9b82a522ed : SUCCESS in 4m 40s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/277bf41dd55f45558d9df2229ef702b2 : SUCCESS in 5m 10s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/eada50cc967f4dfebf0eba5b9a3fe5b3 : FAILURE in 5m 54s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/63b5e1cb31934deea905e3b3b9fc4125 : FAILURE in 2m 03s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/14f6259404c74a318ab91b7ad4fd39c1 : FAILURE in 2m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d058da7677fe457da6231fe4781e2ea3 : SUCCESS in 42m 32s","accounts_in_message":[],"_revision_number":2},{"id":"d7433a1c6e092d23dc4da450e196cfafbc24b7cf","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2022-05-10 00:15:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b124c9412620cd2b20c5ade77445cffc6853fbf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 01:02:27.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/04aadc3fc01a43b0a903cda3303394d4 : FAILURE in 39m 57s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d570c0c263644d0a8b6fc29bd4ad05aa : FAILURE in 11m 43s (non-voting)\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/aa67a8778fb844cba8b349693ca38302 : FAILURE in 14m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ecffb62741aa48ccbe86c7133385ae6f : FAILURE in 21m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89966a67d66e433aaecd7543282cfabc : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75329e37b41a4ff5bfeb853701cf0e46 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f08ba04ce5f44d0af8210440349ee4e : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2952e60a006f43899bbf68cf722576b0 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9cc4f780f874cab8aeb8724381c8179 : SUCCESS in 7m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5a3e2601db514e49b2b13a65628a5d94 : SUCCESS in 5m 07s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/3d5c0ae3e3704ec491c83465a102e77c : SUCCESS in 4m 59s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/a986baa32a7d4c9894070e9de0c1e2f7 : SUCCESS in 12m 06s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/64e9d282792d40c69cefe38ff58af877 : FAILURE in 1m 59s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/171588e720d641a1abad54f39fff4a26 : FAILURE in 2m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/902f005ea71445cea9cff40cf68eeeaa : FAILURE in 10m 41s","accounts_in_message":[],"_revision_number":2},{"id":"eafa5e579d5050e9eb44730e35fc32c01e612584","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2022-05-10 05:34:56.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"5fd6f2c941a1a2b25eed13339277fb792d27ed91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 06:41:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/ba1c8387855647578d3b8322eb271328 : SUCCESS in 58m 10s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6b42e8fef4874721af112ce486f39e99 : FAILURE in 16m 00s (non-voting)\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/1c55870a336e464e85f176afa485f988 : SUCCESS in 58m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b478aba36e434f58b9b16a9aff53ef86 : SUCCESS in 47m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d798b831fa0436a939a877f50edf376 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b67441c15244895819e75004abf109c : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0999c455294446fa8b5a97262f730b3b : SUCCESS in 5m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/22dfc3f85b9b4dab91f865a0c31f8d10 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ccc178f93114b39bd5fac65dcdbb787 : SUCCESS in 7m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bebd7217193f49c29e1b363bd5afc2e1 : SUCCESS in 5m 40s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/2fd6f4bc27904bffa8ca854ec7241d4f : SUCCESS in 6m 45s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/b0f0ce5f16b947b5a4a103023b0ac4f2 : SUCCESS in 12m 55s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/81e27a95c5c14354a95f7a7670224abe : FAILURE in 2m 56s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/16e0656beca541e091af01151b7d3b77 : FAILURE in 2m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a6804ab31134b03b4ddc10a31ee017c : SUCCESS in 44m 07s","accounts_in_message":[],"_revision_number":3},{"id":"31cfc35eb31e75ac3fdb912f97aaac0533e570bb","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2022-05-10 17:35:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c1be0498ae06efb53dd70c0a17a7ad0c5bf20e66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2022-05-13 20:05:37.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"d03ab22525f9d973b3b73d640b1a75796b9a961b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 21:29:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/4d38a4abbf234dfa9627f071f73f5247 : SUCCESS in 53m 53s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3162396ca3f5485c96e79375b386a451 : FAILURE in 15m 15s (non-voting)\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/2e16db6272b74d5d83ddc4b24187a4e0 : SUCCESS in 1h 17m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c2372694d29c4d31a822bcbfbbe24957 : SUCCESS in 56m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e092ab41f054a999b2ed0e20ca66310 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caeccb2ac21246d493bb62e27a197efe : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/104c8547de3d42fc90246c558f95a308 : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/768cacfb9c794b5e829f55a18199dae0 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1175b6113704aea803c15ec94730ad3 : SUCCESS in 6m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6b36f5c0c03048778d61b1eea9f3cbe9 : SUCCESS in 4m 51s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/f20c101d3c4e48c9a4667a5af3a653ab : SUCCESS in 5m 43s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/1ce8a0babf824ebf8726955df40b56dc : SUCCESS in 11m 54s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/b060862986084f87a5fe4ae6fdae00e6 : FAILURE in 2m 15s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/bc89a062e1204cad98c3780d736c1bbd : FAILURE in 2m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c371a12edb64d5fbacb9ebfd48c15b5 : SUCCESS in 39m 41s","accounts_in_message":[],"_revision_number":4},{"id":"fae423552192185ed864ca6c6f8289157a253182","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-12-20 13:14:13.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"5805febf2fd722aad68bb65cdbfd91ea7d83d0d0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-12-20 13:14:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"67b51c24bc49a104342836c497593480108ec912","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2023-01-05 02:16:29.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"ca83886660a26c41af27c3ec8fa6c6648dedbb69","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2023-01-05 02:18:02.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"02ccd17da7e43766a9f51604273b4785bc280fb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 03:36:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ffa927d5f20482e8ea58feb4b2453db\n\n- grenade https://zuul.opendev.org/t/openstack/build/e9e8b3195bfd483a94428d29c11ce367 : SUCCESS in 1h 11m 37s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/e05e556da6ff435ebafb0550227f77a6 : SUCCESS in 1h 08m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7c582c33a664766b4978de1a6bc095a : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2290295b311549609cc68748694e300f : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d57f506e110f4fcfb45cc7dac5e050b4 : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cbd626cbe57e4240b44dfa6613d42a68 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e1d1fcd4de74f8fbbcd61140cdb7e08 : SUCCESS in 6m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/36a5a3daca6d40708ccf5dd5cd06fa84 : SUCCESS in 4m 53s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/4e3b8d0118584c53bb94dfa4f1992113 : SUCCESS in 4m 50s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/8bdb97f5062144ac8d2503919afc5d11 : SUCCESS in 12m 38s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/dd9a6254a9db4eb3b73c2ca6ef7a4caa : FAILURE in 1m 35s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/869199201d73446da26a96c42bf4b3b6 : FAILURE in 1m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46819ffb361f410599df7824bedca987 : SUCCESS in 57m 13s","accounts_in_message":[],"_revision_number":5},{"id":"ae6af950a4ea6fc97254a8f61b281bafe8f248f1","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-01-10 12:38:36.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"78677fdea88af6aba5fc75816c74a2e4dbfbdf74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 12:39:17.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f9920b8de1c62ad42b68be59be3bb5fba63b04e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 14:04:08.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8decc141be94912946f1886b5140b62\n\n- grenade https://zuul.opendev.org/t/openstack/build/a5f3d618ca61471b9b4cb4ae91175d17 : SUCCESS in 1h 07m 21s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/d5317e9e69e743d384e13666de6ea926 : SUCCESS in 1h 14m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bf4fa5e954a4a6dbf9dda3f1202d5c5 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/349219e6d1a946cc82c977e21219b26c : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6120cd94a4b6484cbff40992858db38a : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/722f85a9d3b3416a8b4c1d88628482c6 : SUCCESS in 8m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8e4d02a2f1dd43309a6aa137b067a97e : SUCCESS in 6m 02s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/808cec6629f1405695223367f9f1aa3d : SUCCESS in 6m 18s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/385ae1b6e37e4bbe9a295043fb8ec465 : SUCCESS in 12m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89df7deb827c400db2b6f02da5eab51d : SUCCESS in 46m 25s","accounts_in_message":[],"_revision_number":5},{"id":"89d6a48b75edbe5e2f2b0d065bc2e929d9910342","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 14:04:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"9dfbe9dea2b4082eb9e2b94cd460383a39d237a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 14:05:41.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d232263149d4e33bbbd155dc3db666f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f033760afc8a478995c1218f93b8e1d0 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5a865abc2545544870ad972f70cd54ebd14c19a8","revisions":{"d8bb1b06dd69834523ebec26f418ad08d982aecc":{"kind":"REWORK","_number":1,"created":"2022-05-05 16:30:17.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/03/840703/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/03/840703/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/03/840703/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/03/840703/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/03/840703/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/03/840703/1"}}},"commit":{"parents":[{"commit":"de030ef7a78b1b9ee2da75552d1edc829f830aa4","subject":"Fix Placement Doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/de030ef7a78b1b9ee2da75552d1edc829f830aa4"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2020-11-17 08:19:44.000000000","tz":180},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2022-05-05 16:30:10.000000000","tz":0},"subject":"placement-status: check only consumers in allocation table","message":"placement-status: check only consumers in allocation table\n\nInitially, _check_incomplete_consumers was intended to find the case\nwhen consumers are mentioned in allocation but do not exist in the\nconsumer table. Due to a bug it\u0027s possible to have rows in the\nconsumer table those are not in the allocation table. The second case\nit seems was not expected and it can be tricky to automate a cleanup\nwith zero touch upgrade.\n\nThis patch modifies the check such way it counts only consumers\npresented in the allocation table.\n\nRelated-bug: #1780799\nChange-Id: I569043dd9f1c964faf7d4fcbb901a93f9569bf3d\n(cherry picked from commit 3253e2efc05836b0b5cefafbf2b2656e9e1a50c7)\n(cherry picked from commit 69a0fb8bae4bc2538df266d8cbd788ac2ad61094)\n(cherry picked from commit abbec22662a8ea949f252acee4eb5f8f4d2242a4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d8bb1b06dd69834523ebec26f418ad08d982aecc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d8bb1b06dd69834523ebec26f418ad08d982aecc"}]},"branch":"refs/heads/stable/ussuri"},"6e4cf99e596a3a7c5afde048d8daf0bcfe1976d7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-05-05 22:10:58.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/03/840703/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/03/840703/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/03/840703/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/03/840703/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/03/840703/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/03/840703/2"}}},"commit":{"parents":[{"commit":"445a7b544b5fce45a2e79046df1fdc2548e6db9e","subject":"Use \u0027functional-without-sample-db-tests\u0027 tox env for placement nova job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/445a7b544b5fce45a2e79046df1fdc2548e6db9e"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2020-11-17 08:19:44.000000000","tz":180},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2022-05-05 22:10:23.000000000","tz":0},"subject":"placement-status: check only consumers in allocation table","message":"placement-status: check only consumers in allocation table\n\nInitially, _check_incomplete_consumers was intended to find the case\nwhen consumers are mentioned in allocation but do not exist in the\nconsumer table. Due to a bug it\u0027s possible to have rows in the\nconsumer table those are not in the allocation table. The second case\nit seems was not expected and it can be tricky to automate a cleanup\nwith zero touch upgrade.\n\nThis patch modifies the check such way it counts only consumers\npresented in the allocation table.\n\nRelated-bug: #1780799\nChange-Id: I569043dd9f1c964faf7d4fcbb901a93f9569bf3d\n(cherry picked from commit 3253e2efc05836b0b5cefafbf2b2656e9e1a50c7)\n(cherry picked from commit 69a0fb8bae4bc2538df266d8cbd788ac2ad61094)\n(cherry picked from commit abbec22662a8ea949f252acee4eb5f8f4d2242a4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/6e4cf99e596a3a7c5afde048d8daf0bcfe1976d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/6e4cf99e596a3a7c5afde048d8daf0bcfe1976d7"}]},"branch":"refs/heads/stable/ussuri"},"244be37185df74d6a753af5558b4226cb447486a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-05-10 05:34:56.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/03/840703/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/03/840703/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/03/840703/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/03/840703/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/03/840703/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/03/840703/3"}}},"commit":{"parents":[{"commit":"4fd1d0cde3d2f5c4286d5b6a124c5c26c312da90","subject":"Use \u0027functional-without-sample-db-tests\u0027 tox env for placement nova job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/4fd1d0cde3d2f5c4286d5b6a124c5c26c312da90"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2020-11-17 08:19:44.000000000","tz":180},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2022-05-10 05:34:45.000000000","tz":0},"subject":"placement-status: check only consumers in allocation table","message":"placement-status: check only consumers in allocation table\n\nInitially, _check_incomplete_consumers was intended to find the case\nwhen consumers are mentioned in allocation but do not exist in the\nconsumer table. Due to a bug it\u0027s possible to have rows in the\nconsumer table those are not in the allocation table. The second case\nit seems was not expected and it can be tricky to automate a cleanup\nwith zero touch upgrade.\n\nThis patch modifies the check such way it counts only consumers\npresented in the allocation table.\n\nRelated-bug: #1780799\nChange-Id: I569043dd9f1c964faf7d4fcbb901a93f9569bf3d\n(cherry picked from commit 3253e2efc05836b0b5cefafbf2b2656e9e1a50c7)\n(cherry picked from commit 69a0fb8bae4bc2538df266d8cbd788ac2ad61094)\n(cherry picked from commit abbec22662a8ea949f252acee4eb5f8f4d2242a4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/244be37185df74d6a753af5558b4226cb447486a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/244be37185df74d6a753af5558b4226cb447486a"}]},"branch":"refs/heads/stable/ussuri"},"ca8579dc863f65cfeae1601a009270df5b23e5b2":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-05-13 20:05:37.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/03/840703/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/03/840703/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/03/840703/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/03/840703/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/03/840703/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/03/840703/4"}}},"commit":{"parents":[{"commit":"4fd1d0cde3d2f5c4286d5b6a124c5c26c312da90","subject":"Use \u0027functional-without-sample-db-tests\u0027 tox env for placement nova job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/4fd1d0cde3d2f5c4286d5b6a124c5c26c312da90"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2020-11-17 08:19:44.000000000","tz":180},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2022-05-13 20:05:19.000000000","tz":0},"subject":"placement-status: check only consumers in allocation table","message":"placement-status: check only consumers in allocation table\n\nInitially, _check_incomplete_consumers was intended to find the case\nwhen consumers are mentioned in allocation but do not exist in the\nconsumer table. Due to a bug it\u0027s possible to have rows in the\nconsumer table those are not in the allocation table. The second case\nit seems was not expected and it can be tricky to automate a cleanup\nwith zero touch upgrade.\n\nThis patch modifies the check such way it counts only consumers\npresented in the allocation table.\n\nConflicts:\n    placement/cmd/status.py\n\nNOTE(melwitt): The conflict is because change\nIe0bee79c1ee185c0f68fa170c668395d9058c56d (Fix SQL query counting the\nnumber of individual consumers having allocations by only selecting the\naggregated consumer_id column.) is not in Wallaby.\n\nRelated-bug: #1780799\nChange-Id: I569043dd9f1c964faf7d4fcbb901a93f9569bf3d\n(cherry picked from commit 3253e2efc05836b0b5cefafbf2b2656e9e1a50c7)\n(cherry picked from commit 7f66845b6382513ae291914ea02787e9776505a0)\n(cherry picked from commit 93bd7f23ad6e993185e32f038db90bb81e8746e2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ca8579dc863f65cfeae1601a009270df5b23e5b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ca8579dc863f65cfeae1601a009270df5b23e5b2"}]},"branch":"refs/heads/stable/ussuri"},"5a865abc2545544870ad972f70cd54ebd14c19a8":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2023-01-05 02:16:29.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/03/840703/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/03/840703/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/03/840703/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/03/840703/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/03/840703/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/03/840703/5"}}},"commit":{"parents":[{"commit":"4fd1d0cde3d2f5c4286d5b6a124c5c26c312da90","subject":"Use \u0027functional-without-sample-db-tests\u0027 tox env for placement nova job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/4fd1d0cde3d2f5c4286d5b6a124c5c26c312da90"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2020-11-17 08:19:44.000000000","tz":180},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2023-01-05 02:16:22.000000000","tz":0},"subject":"placement-status: check only consumers in allocation table","message":"placement-status: check only consumers in allocation table\n\nInitially, _check_incomplete_consumers was intended to find the case\nwhen consumers are mentioned in allocation but do not exist in the\nconsumer table. Due to a bug it\u0027s possible to have rows in the\nconsumer table those are not in the allocation table. The second case\nit seems was not expected and it can be tricky to automate a cleanup\nwith zero touch upgrade.\n\nThis patch modifies the check such way it counts only consumers\npresented in the allocation table.\n\nRelated-bug: #1780799\nChange-Id: I569043dd9f1c964faf7d4fcbb901a93f9569bf3d\n(cherry picked from commit 3253e2efc05836b0b5cefafbf2b2656e9e1a50c7)\n(cherry picked from commit 7f66845b6382513ae291914ea02787e9776505a0)\n(cherry picked from commit 93bd7f23ad6e993185e32f038db90bb81e8746e2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/5a865abc2545544870ad972f70cd54ebd14c19a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/5a865abc2545544870ad972f70cd54ebd14c19a8"}]},"branch":"refs/heads/stable/ussuri","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
