)]}'
{"id":"openstack%2Fplacement~840718","triplet_id":"openstack%2Fplacement~stable%2Fwallaby~If7e6060c867df21360a1f588ca872d65d75eb3c1","project":"openstack/placement","branch":"stable/wallaby","topic":"bug/1780799","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2022-05-24 18:37:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If7e6060c867df21360a1f588ca872d65d75eb3c1","subject":"Use \u0027functional-without-sample-db-tests\u0027 tox env for placement nova job","status":"MERGED","created":"2022-05-05 18:38:52.000000000","updated":"2022-05-24 18:40:05.000000000","submitted":"2022-05-24 18:37:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"840718-bug/1780799","meta_rev_id":"21b557769ac2c1309d1e687e157fe50a8cf8041d","_number":840718,"virtual_id_number":840718,"owner":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-24 18:37:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2022-05-24 17:20:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2022-05-24 16:42:12.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2022-05-24 17:20:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-05 18:38:52.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2022-05-05 19:52:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-24 16:42:12.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2022-05-24 17:20:09.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"8d92bf493a37e0251db8ef510ef1ef9e7064ccf4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2022-05-05 18:38:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1ab28b84197204411c5aada19afaa94a9213b7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 19:52:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/37784c692b0643879cf16971a135f2c8 : SUCCESS in 50m 13s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a6424987f1af4ea5b1d2deb63352a6cd : FAILURE in 51m 05s (non-voting)\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/7e2a16ab15f9481688503aff094d562f : SUCCESS in 53m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2d1b9fa9aa0e401b8c443d96dbb8cbc1 : SUCCESS in 56m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fdeb9a5c9c8c40178cd498d3c07ee8b4 : SUCCESS in 5m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c31ea03106e84081a7639728f48cb6f9 : FAILURE in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eec0f5527e8346f8b81b5c7a71f1953d : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1c566b6e7bc49b8aca2f591ec8658e7 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df03a690bad3490f841f59fb99dab207 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/909d426c039e416f81aab00559551a5a : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e5c7970792e4e31abab8c66992f5661 : SUCCESS in 7m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aa9aefb28cfb4899b4080125589248b4 : SUCCESS in 5m 20s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/8e71174579014150b1c9419d08b2c362 : SUCCESS in 5m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cafd13821d1d479cb9ce6e00bc4d3639 : SUCCESS in 5m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ff1634440318446198d7bc157703f2c9 : SUCCESS in 6m 25s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c4f6fb0278f84457aba303203ff54d90 : SUCCESS in 14m 02s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/7111a00673e543f7b71491d1fb524a12 : TIMED_OUT in 1h 00m 56s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/a1a7b62471c44037aa2a76353df851c9 : FAILURE in 10m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b2bdf43915b4cd09724c1b836117b8b : SUCCESS in 52m 03s","accounts_in_message":[],"_revision_number":1},{"id":"f0f01a8436c14593b26f46003052d3b1b3ebb34c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2022-05-05 21:30:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dcb523589ffb1d3f08aee736770e90414a7f3e95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 22:49:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/506e8b3e6e0c4e42a1ba183cd11e56dd : SUCCESS in 4m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/f080335f22cf4992bc351b73b10a23da : SUCCESS in 1h 07m 04s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/564a3015525c4d3a8d1e08a2f450c0d5 : FAILURE in 49m 08s (non-voting)\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/b3574e7f4f414068a1a4e40c19842496 : SUCCESS in 1h 13m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e7b8c37c17e4b08b0b11559dee42a4f : SUCCESS in 59m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/326ea429c2064be5900b52aabeb1c0ee : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e406df949594c258aca0dde4e64316a : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2bc6d9f106d04cb89db21f7abe86eb95 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4cedd43a02d34a3f9e452a5e32dbf4b6 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/124c7242e8ee47c99defc89f7da11bc6 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/566c956f84b64e6bac615e2612f6120c : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd0d1836912d46989740a7f58adcadf7 : SUCCESS in 4m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7c748f3fb61b44eb95f701ba410bae94 : SUCCESS in 5m 25s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/120ff7030c44417fad200ec4fb9e5a8e : SUCCESS in 5m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ba8014e10ce44919955b02fbacba03c4 : SUCCESS in 5m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c64a6ff5135c41258a71498681609d09 : SUCCESS in 6m 30s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e0c8408c9cb34cd6b5c909183d803348 : SUCCESS in 14m 17s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/b02d16e016484c9789c54d2141a97061 : SUCCESS in 21m 09s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/b0336692bb884798adad2206ca4c2fc6 : SUCCESS in 8m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab3fdfe3194b49589cdb00b2852727c0 : SUCCESS in 50m 44s","accounts_in_message":[],"_revision_number":2},{"id":"711cc325fdff80ee66874426c9c980272fa847d2","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-05-24 16:42:12.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"463db1506f13a88e7210f92b902ef5cbb924ac44","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-05-24 17:20:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"cfb3a3f3b2e67ea56b1511e14e024d3f48cd77fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 17:21:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3b80ad323e402ce59273c51e5488c4f8c3fc4062","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 18:37:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d3c49a865a8844929d99f6e1c28eee74 : SUCCESS in 5m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/d958d28c77c44f84a746a574aa5235d8 : SUCCESS in 59m 23s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/01afd2d930aa4ab9be89cfdfd8e993dd : SUCCESS in 1h 11m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e43d549ce8545cd8be6a0bac014be4b : SUCCESS in 56m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c97e10d6b1945bea36432d765360f59 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e4824fbaf5846afbe5bd6b869adcfdc : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12ef8ea098e04423b14a99862319b347 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af8dfc0da784421aa07ccd7764a6d418 : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ab98e0bc38b49dbbb6b9cb7bd2b8950 : SUCCESS in 6m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d56d510059ff4b26a7017931da7c0679 : SUCCESS in 6m 01s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/e1cf682e14f14fb9846ad1bcff4e4324 : SUCCESS in 6m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/61e42a2c6b934285af42ddc621ad9365 : SUCCESS in 6m 20s\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c46bffe5d820472aa3299928e0dc6cbd : SUCCESS in 15m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/daa995df4ef6499c89c7cb0053f24260 : SUCCESS in 1h 08m 46s","accounts_in_message":[],"_revision_number":2},{"id":"b47ba3c3f41e77526590527dc5dcbcf73537cbdc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 18:37:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"21b557769ac2c1309d1e687e157fe50a8cf8041d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 18:40:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5059a8528c1040d4962e345cb66fdf49 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91d3bfbb31ee489d9a2ec7b017970cbb : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"69dcdcb2b7df0a81addec55804ce91c724ca0f54","revisions":{"cf1f54df20f8ef66a8e3f1316d222cee8f5eb8b5":{"kind":"REWORK","_number":1,"created":"2022-05-05 18:38:52.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/18/840718/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/18/840718/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/18/840718/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/18/840718/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/18/840718/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/18/840718/1"}}},"commit":{"parents":[{"commit":"adf525a8f176fe37222d08478c9fe08db823db98","subject":"Make sure the policy upgrade check get a valid config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/adf525a8f176fe37222d08478c9fe08db823db98"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-10-12 17:33:15.000000000","tz":-300},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2022-05-05 18:38:35.000000000","tz":0},"subject":"Use \u0027functional-without-sample-db-tests\u0027 tox env for placement nova job","message":"Use \u0027functional-without-sample-db-tests\u0027 tox env for placement nova job\n\nNova defines (depends-on) a new tox env \u0027functional-without-sample-db-tests\u0027\nto run the nova functional test by exlcuding the api|notification\nsample_tests, and db-related tests.\n\nWe can use that to fix the error  are facing currently for this job\n- https://zuul.openstack.org/build/a8c186b2c7124856ae32477f10e2b9a4\n\nDepends-On: https://review.opendev.org/c/openstack/nova/+/840717\nChange-Id: If7e6060c867df21360a1f588ca872d65d75eb3c1\n(cherry picked from commit 6d49b6df1d3f62e2003c55509875380148281a79)\n(cherry picked from commit 7422ed4d7adb26868602a3f4652119ff0392d298)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/cf1f54df20f8ef66a8e3f1316d222cee8f5eb8b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/cf1f54df20f8ef66a8e3f1316d222cee8f5eb8b5"}]},"branch":"refs/heads/stable/wallaby"},"69dcdcb2b7df0a81addec55804ce91c724ca0f54":{"kind":"REWORK","_number":2,"created":"2022-05-05 21:30:05.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/18/840718/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/18/840718/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/18/840718/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/18/840718/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/18/840718/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/18/840718/2"}}},"commit":{"parents":[{"commit":"adf525a8f176fe37222d08478c9fe08db823db98","subject":"Make sure the policy upgrade check get a valid config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/adf525a8f176fe37222d08478c9fe08db823db98"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-10-12 17:33:15.000000000","tz":-300},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2022-05-05 21:20:00.000000000","tz":0},"subject":"Use \u0027functional-without-sample-db-tests\u0027 tox env for placement nova job","message":"Use \u0027functional-without-sample-db-tests\u0027 tox env for placement nova job\n\nNOTE(melwitt): This is a combination of 2 commits needed to fix the\ngate, a chicken and egg problem where each fix depends on the other.\n\nNova defines (depends-on) a new tox env \u0027functional-without-sample-db-tests\u0027\nto run the nova functional test by exlcuding the api|notification\nsample_tests, and db-related tests.\n\nWe can use that to fix the error  are facing currently for this job\n- https://zuul.openstack.org/build/a8c186b2c7124856ae32477f10e2b9a4\n\nDepends-On: https://review.opendev.org/c/openstack/nova/+/840717\n\nDrop lower-constraints.txt and its testing\n\nAs discussed in TC PTG[1] and TC resolution[2], we are\ndropping the lower-constraints.txt file and its testing.\nWe will keep lower bounds in the requirements.txt file but\nwith a note that these are not tested lower bounds and we\ntry our best to keep them updated.\n\nConflicts:\n    .zuul.yaml\n    lower-constraints.txt\n\nNOTE(melwitt): conflict is due to branch specific settings (job\ntemplate names, lower constraints changes).\n\n[1] https://etherpad.opendev.org/p/tc-zed-ptg#L326\n[2] https://governance.openstack.org/tc/resolutions/20220414-drop-lower-constraints.html#proposal\n\nChange-Id: Ieba7daf39fa3323e8c9a7396747449f24189fcd5\n(cherry picked from commit 03d567928e31d3dc85d4dd3f5617785e7380b6b1)\n(cherry picked from commit 4d3df47ee3e394e3178d58c15306620809ad2806)\n(cherry picked from commit f96984abd48f4d938d0fff0bab9b12ed933e320e)\n\nChange-Id: If7e6060c867df21360a1f588ca872d65d75eb3c1\n(cherry picked from commit 6d49b6df1d3f62e2003c55509875380148281a79)\n(cherry picked from commit 7422ed4d7adb26868602a3f4652119ff0392d298)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/69dcdcb2b7df0a81addec55804ce91c724ca0f54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/69dcdcb2b7df0a81addec55804ce91c724ca0f54"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
