)]}'
{"id":"openstack%2Fplacement~840778","triplet_id":"openstack%2Fplacement~stable%2Ftrain~If7e6060c867df21360a1f588ca872d65d75eb3c1","project":"openstack/placement","branch":"stable/train","topic":"bug/1780799","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2023-09-01 17:08:02.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If7e6060c867df21360a1f588ca872d65d75eb3c1","subject":"Use \u0027functional-without-sample-db-tests\u0027 tox env for placement nova job","status":"ABANDONED","created":"2022-05-05 22:25:46.000000000","updated":"2023-09-01 17:08:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e2eb1993f87cf10bffad2bc71173ce501c19210f","_number":840778,"virtual_id_number":840778,"owner":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-05-05 23:42:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-05 22:25:46.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2022-05-05 23:42:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b3245c62269a3b8eeaf2f688dc9e645debe30359","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2022-05-05 22:25:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a21ffddb54d61542545f4e89e3a0ed01a62e515","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 23:42:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/adc7d0215e4044669de80373e97b6295 : SUCCESS in 1h 07m 36s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/edde73dbfba74839b0d5327c0ad5024f : FAILURE in 22m 43s (non-voting)\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/b5dba7ea0bf744dbad6b00584fe28941 : SUCCESS in 1h 11m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ce941d04ab64acfbf760afa942a014f : SUCCESS in 59m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6817666b0a0b440bb50d26265d43d44c : SUCCESS in 7m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b1b44c680e94d57a083779c8e52afc2 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/315477a4f66646ca99537019dcc87da0 : SUCCESS in 4m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fe2fa4e1ad874d03b8c058de09bc35fc : FAILURE in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/145d9e71375c4ca4a1ca77cf0f2b0d18 : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0fc3c4ea85c443fbbc8736f37d515e93 : FAILURE in 4m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6564647264b4499ea1750d7916f1c2fd : SUCCESS in 5m 01s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/af8b9860db7b4b689fd3ab05dc7a4a66 : FAILURE in 4m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8ef3ac6b4f6d4be4a8f718930a318dc0 : SUCCESS in 5m 10s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/196eed698a2f4707ada5313e07582fe4 : FAILURE in 4m 57s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/3eac6f08d67f4798b94422ee81e32598 : FAILURE in 5m 32s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/3e39eda67fe84469af08489b471ef40a : FAILURE in 2m 13s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/7e005a6accbe49d9b4b55b583ba3f54d : FAILURE in 2m 10s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/aea0695bd38b4721a46ee45102be9db6 : SUCCESS in 1h 02m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/309a13696d1d4a50ab53bc43891abb34 : SUCCESS in 58m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e2eb1993f87cf10bffad2bc71173ce501c19210f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-01 17:08:02.000000000","message":"Abandoned\n\nstable/train branch of nova projects\u0027 have been tagged as End of Life. All open patches have to be abandoned in order to be able to delete the branch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"42b153a52ca4e279d7b0e45ef42df442622fce0e","revisions":{"42b153a52ca4e279d7b0e45ef42df442622fce0e":{"kind":"REWORK","_number":1,"created":"2022-05-05 22:25:46.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/78/840778/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/78/840778/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/78/840778/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/78/840778/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/78/840778/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/78/840778/1"}}},"commit":{"parents":[{"commit":"ff55034f10ab0b45ebcc455434fa491efcd3eb7b","subject":"Fix Placement Doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ff55034f10ab0b45ebcc455434fa491efcd3eb7b"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-10-12 17:33:15.000000000","tz":-300},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2022-05-05 22:25:37.000000000","tz":0},"subject":"Use \u0027functional-without-sample-db-tests\u0027 tox env for placement nova job","message":"Use \u0027functional-without-sample-db-tests\u0027 tox env for placement nova job\n\nNOTE(melwitt): This is a combination of 2 commits needed to fix the\ngate, a chicken and egg problem where each fix depends on the other.\n\nNova defines (depends-on) a new tox env \u0027functional-without-sample-db-tests\u0027\nto run the nova functional test by exlcuding the api|notification\nsample_tests, and db-related tests.\n\nWe can use that to fix the error  are facing currently for this job\n- https://zuul.openstack.org/build/a8c186b2c7124856ae32477f10e2b9a4\n\nDepends-On: https://review.opendev.org/c/openstack/nova/+/840777\n\nDrop lower-constraints.txt and its testing\n\nAs discussed in TC PTG[1] and TC resolution[2], we are\ndropping the lower-constraints.txt file and its testing.\nWe will keep lower bounds in the requirements.txt file but\nwith a note that these are not tested lower bounds and we\ntry our best to keep them updated.\n\nConflicts:\n    .zuul.yaml\n    lower-constraints.txt\n\nNOTE(melwitt): conflict is due to branch specific settings (job\ntemplate names, lower constraints changes).\n\n[1] https://etherpad.opendev.org/p/tc-zed-ptg#L326\n[2] https://governance.openstack.org/tc/resolutions/20220414-drop-lower-constraints.html#proposal\n\nChange-Id: Ieba7daf39fa3323e8c9a7396747449f24189fcd5\n(cherry picked from commit 03d567928e31d3dc85d4dd3f5617785e7380b6b1)\n(cherry picked from commit 4d3df47ee3e394e3178d58c15306620809ad2806)\n(cherry picked from commit f96984abd48f4d938d0fff0bab9b12ed933e320e)\n\nChange-Id: If7e6060c867df21360a1f588ca872d65d75eb3c1\n(cherry picked from commit 6d49b6df1d3f62e2003c55509875380148281a79)\n(cherry picked from commit 7422ed4d7adb26868602a3f4652119ff0392d298)\n(cherry picked from commit 69dcdcb2b7df0a81addec55804ce91c724ca0f54)\n(cherry picked from commit 503b1f27cefeadf127d3f9f274901eba52f67c01)\n(cherry picked from commit 445a7b544b5fce45a2e79046df1fdc2548e6db9e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/42b153a52ca4e279d7b0e45ef42df442622fce0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/42b153a52ca4e279d7b0e45ef42df442622fce0e"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
