)]}'
{"id":"openstack%2Fplacement~858927","triplet_id":"openstack%2Fplacement~master~I7fa25fea03f0c9c645a1eb10b22f058ca548d2fb","project":"openstack/placement","branch":"master","topic":"fix-bindeps","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2022-09-22 22:23:57.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-09-22 14:30:54.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I7fa25fea03f0c9c645a1eb10b22f058ca548d2fb","subject":"update bindep for ubuntu 22.04","status":"MERGED","created":"2022-09-22 13:19:38.000000000","updated":"2022-09-22 22:25:56.000000000","submitted":"2022-09-22 22:23:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858927-fix-bindeps","meta_rev_id":"43114931779930ad797b099fa698a5f84952d18d","_number":858927,"virtual_id_number":858927,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-22 22:23:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"recommended":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"all":[{"value":1,"date":"2022-09-22 19:19:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":2,"date":"2022-09-22 14:06:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2022-09-22 14:59:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-22 15:10:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":1,"date":"2022-09-22 14:57:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-22 15:10:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-22 13:45:40.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-09-22 14:59:08.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-09-22 15:10:35.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2022-09-22 15:49:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-22 19:19:20.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"}],"messages":[{"id":"67d5535c11155c5564cd0f600ad34fa4861118ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-09-22 13:19:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43306fbcf8011a6314fbeffc558da9c40a360433","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-22 13:45:40.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cd35a84bf797b48140a270303b5fba067fe86aa2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-09-22 13:58:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fcf4223e6c902293eea6d67cf10bc7fceb04873f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-22 14:06:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5edd3ca6fb4ceebfbfb3230ced0775a719f544ea","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-22 14:07:58.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a1738c849e2b7cfdece1d9012f89811f1861ea47","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-09-22 14:23:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"20d8162cb40aace0b3884b22de5b1a372eefa509","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-22 14:30:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c4d22985b38bc8789f446f1583ed93df277625b2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-22 14:56:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"49c8daa6d8434113993a841587822fb62096ac83","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-22 14:57:20.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"09ae96b4c97c6cf35759ea2c034598b549699f5e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-09-22 14:59:08.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"367eb51de4a5925fb96de7bff70a62b767297cdb","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-09-22 15:10:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f57b09faa4ad9190809a045f41b5ccb446f349a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 15:49:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7292bd3668514e98a3f1d893c8ee8117\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8ad5f0a5e7ec47a5a52264cabb971f93 : SUCCESS in 5m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/b969939e316b4faa9c5d792916807b63 : SUCCESS in 1h 02m 08s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/95f61f4929694142bfc0de61b24a2eab : SUCCESS in 1h 44m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d6152132612746718a7600267d178469 : SUCCESS in 1h 00m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d960c70a0ced4cd79dc0461a083b2580 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57207b8e1013413d807191b9b0c2a504 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5811cd1b47dc4bfcbd21700d6bbc299a : SUCCESS in 7m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e86bda711ecf4b0bbf4d3cc0f7c180ef : SUCCESS in 6m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4e0a312567b043f1a790335df1fa18c2 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a613eef8f4e412db8f1474ac0912e2c : SUCCESS in 10m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/764cc24f8bd145fea98bc311d9e1c92e : SUCCESS in 5m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0014037604544a9399d01bae9df007d4 : SUCCESS in 8m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7e2bfdd6b52c4aedad2f280e1f984797 : SUCCESS in 5m 49s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f823310af7ef410b98ad380cabbd4526 : SUCCESS in 20m 33s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/d3e31e283ec54745a969e273073fd923 : SUCCESS in 28m 00s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/b0eaf00c79ef4e7b80bfd409dbdc79bb : SUCCESS in 9m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9a63968f9f94fbd9f76f240b54e2056 : SUCCESS in 59m 47s","accounts_in_message":[],"_revision_number":2},{"id":"6430fd92ac5a531262e47656723fee25b7d96cd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 15:49:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"349329cfabe6250bc2592b2f853ae46b6cb48dd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 17:08:10.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd15b3e9787a4c8bbaf80c9b16696e86\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0e087a5cbdf44ad0a5c1ad98ec388b86 : SUCCESS in 5m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/8e372ae6c435452ab749945ecabded7c : SUCCESS in 1h 04m 32s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/87924027184143ada1ce1ab7dfa9e635 : SUCCESS in 1h 17m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f5912d07adc41028d0fbcbb0079f859 : SUCCESS in 1h 00m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e8ddaead42244b180e71567a62d51cf : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a1f36536d714965a76ec675e35594ea : SUCCESS in 8m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/073449fa0b084aa396ba901b3eca5838 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/124bc034c3954685b8546e6235f85d4a : SUCCESS in 11m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ad0bee6ca7047ba87e30d63824ae576 : SUCCESS in 5m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4941d345d9364c9791e1392e093de92a : SUCCESS in 6m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1f7daba84eca49c68ba0ddb1e440ea8f : SUCCESS in 6m 31s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2db9d4d404e64fe9865c44304e26644d : FAILURE in 17m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d97a960f843b4594ab450f9ef6fcd7c0 : SUCCESS in 1h 00m 25s","accounts_in_message":[],"_revision_number":2},{"id":"48f21c4ccdb725944f267678c8a32441183a6c4c","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2022-09-22 19:19:20.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6038ca32c7cc48c92766b10bb9b9089678c02310","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 20:36:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ce5a1dd5776461bbb9b81db62c80db4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0857ab592a544315912df8c53947d911 : SUCCESS in 6m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/cc3e5c1dd1494ea7a29b25b1f197e238 : SUCCESS in 43m 57s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/1f1d081206f34fe1879c9316859cf282 : SUCCESS in 1h 15m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe3365741db647b98d2f89980b188592 : SUCCESS in 52m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb74ff0122474e36930711387d5a2baf : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/700ccb41835b49ff876b755d7dcb67ca : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00741fe766774223bb7ec9905d255431 : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7eb2a141861d4050ace13e9e38626a47 : SUCCESS in 5m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ada83948905425d9fcecd078dece66c : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb887df834024a77aca38885899734ee : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b77fbdf59263443d8a1f0d5db9fb9ea1 : SUCCESS in 6m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9f0f76c86dce4aa4bb31964d800c6e10 : SUCCESS in 6m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cecf59f8cf47427ea57c5a0175a14309 : SUCCESS in 6m 09s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d9b6bac5269c44e5aeddacbd909dbd6c : SUCCESS in 17m 44s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/72493b96b65347e3b07f6abd01ab5525 : SUCCESS in 17m 23s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/e00525502c144f578ef224c708f643a4 : SUCCESS in 9m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61e09fc336aa4748b747a507f06966d9 : SUCCESS in 56m 48s","accounts_in_message":[],"_revision_number":2},{"id":"d4ced17d733be606afedc284b1c6a2f0406516fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 20:36:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bed61315a0af494930aaa43b71bda23a9112e756","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 22:23:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/921aa3649a9d42c08193eb17e1ff82e3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f05dfe65b7734cc29dafd4bb77089488 : SUCCESS in 5m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/9f1d233f513a4965b2fbad91b82313d6 : SUCCESS in 1h 07m 41s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/76fd34d118ec4c4fbabff54e5be90217 : SUCCESS in 1h 40m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f5bbcd569634604bc765b80b6aa4ebd : SUCCESS in 1h 13m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/168a0ba5f2eb4aeabda0dd6c899ac0e7 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b64e66c558ca48d5807a7a55eceacc7d : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/364d26664fae4e3f9dc73cf60bc27ad2 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f8c3ece5cae40b5bc2a8d3b39670a48 : SUCCESS in 8m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c7d1898341542fabb403006349cf51a : SUCCESS in 5m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ea7fac6740fe4ebfaafcb9eaf85dcfbf : SUCCESS in 6m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a152dcebcaaf412f9bad0f9f1c676925 : SUCCESS in 6m 28s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9e30802ed5134adaa71aaff18c740fb9 : SUCCESS in 19m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72152535b44c4df08250d833e24d1075 : SUCCESS in 1h 00m 39s","accounts_in_message":[],"_revision_number":2},{"id":"2eb5df7371f5d88c6492f414cbbc3d9d08634be9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 22:23:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"43114931779930ad797b099fa698a5f84952d18d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 22:25:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/953cbd8b26504ce2a4898444c21c44fc\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/58faf29441574540b954d26e02b70b44 : SUCCESS in 1m 07s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6d950b210c24973bdd60f6f88eee283 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b079acb49dcf4dcb91de92c31014b554 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"92753e4a3a09e68e4149a91d1ee2761b66805901","revisions":{"e688c88ef3cb56ce5f521ddedd0b1a44810f7eaa":{"kind":"REWORK","_number":1,"created":"2022-09-22 13:19:38.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/27/858927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/27/858927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/27/858927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/27/858927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/27/858927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/27/858927/1"}}},"commit":{"parents":[{"commit":"c6b1b92b1dc967c8721cc160e7f7c78a83950459","subject":"Merge \"Update master for stable/zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/c6b1b92b1dc967c8721cc160e7f7c78a83950459"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2022-09-22 13:16:01.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2022-09-22 13:16:01.000000000","tz":60},"subject":"update bindep for ubuntu 22.04","message":"update bindep for ubuntu 22.04\n\nThe python 3.10 tox job runs on ubuntu 22.04 in the A cycle ubuntu 22.04\nwill also become the default for other jobs. python-dev is nolonger packaged\non ubuntu 22.04 so the tox job fails.\n\nThis change sync the python deps to match those use in nova which\nalready works across all tested distros.\n\nChange-Id: I7fa25fea03f0c9c645a1eb10b22f058ca548d2fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e688c88ef3cb56ce5f521ddedd0b1a44810f7eaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e688c88ef3cb56ce5f521ddedd0b1a44810f7eaa"}]},"branch":"refs/heads/master"},"92753e4a3a09e68e4149a91d1ee2761b66805901":{"kind":"REWORK","_number":2,"created":"2022-09-22 13:58:15.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/27/858927/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/27/858927/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/27/858927/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/27/858927/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/27/858927/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/27/858927/2"}}},"commit":{"parents":[{"commit":"c6b1b92b1dc967c8721cc160e7f7c78a83950459","subject":"Merge \"Update master for stable/zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/c6b1b92b1dc967c8721cc160e7f7c78a83950459"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2022-09-22 13:16:01.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2022-09-22 13:58:08.000000000","tz":60},"subject":"update bindep for ubuntu 22.04","message":"update bindep for ubuntu 22.04\n\nThe python 3.10 tox job runs on ubuntu 22.04 in the A cycle ubuntu 22.04\nwill also become the default for other jobs. python-dev is nolonger packaged\non ubuntu 22.04 so the tox job fails.\n\nThis change sync the python deps to match those use in nova which\nalready works across all tested distros.\n\nChange-Id: I7fa25fea03f0c9c645a1eb10b22f058ca548d2fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/92753e4a3a09e68e4149a91d1ee2761b66805901"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/92753e4a3a09e68e4149a91d1ee2761b66805901"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
