)]}'
{"id":"openstack%2Fplacement~864385","triplet_id":"openstack%2Fplacement~master~I12dce16a7f18bb065e1412d2f2d9bd91452d8672","project":"openstack/placement","branch":"master","topic":"policy-defaults-improvement","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2022-11-30 11:58:36.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-11-30 12:18:47.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2022-11-30 08:24:02.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e replied on the change","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2022-11-30 12:18:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I12dce16a7f18bb065e1412d2f2d9bd91452d8672","subject":"Policy defaults improvement spec","status":"MERGED","created":"2022-11-14 05:19:59.000000000","updated":"2022-11-30 12:20:55.000000000","submitted":"2022-11-30 12:18:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864385-policy-defaults-improvement","meta_rev_id":"522242916f6bbbab173b9b610b951631e6890057","_number":864385,"virtual_id_number":864385,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-30 12:18:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2022-11-30 11:58:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-30 08:24:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2022-11-30 11:58:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-14 05:37:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-15 17:22:57.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2022-11-23 08:38:47.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2022-11-23 08:38:47.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"}],"messages":[{"id":"791b2f333994e9ec2a6284798f352390e6b15519","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-14 05:19:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"899525f9138c22b5b6450c9fd3dec5af5a6fe19c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 05:37:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ae88d339ed34e749f2818c6a4d064e7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a56b9546be204fe5ab6b7dd6e78ecc42 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e9354e1e0a24904af9dd08159e5c95d : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":1},{"id":"172c3db545c52ac87a85bda8fe2e217ff72a505c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-14 19:59:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7d49d0043ba4399407580df65e65c1fc971b1fc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-15 16:34:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c6cef019cc6811de164ccacdb04df0e31d5f672e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 16:46:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/646b2df8c484400097eb0446ca745af2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3956c0975f434c248d9ee0000d1533c0 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c15ddfaff5441089e29c394437390e5 : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":2},{"id":"8a41ef6c9614f06835642da44c686b346596b03d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-11-15 17:22:57.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f7c56d435be35965b66e8ec68cf3098330ad06a5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-15 17:40:27.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b0cab314797c1eeabd38e7f9f51fba1db15ec2f1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-15 17:50:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e4b08e9a968d2a55a9e745745a36322cb9cc8c1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-15 17:51:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"309c39f158f4d69a7084b4653e2ace3df2f9d280","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-15 17:51:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5cf36a46f5e3f2341603de8d24e2ac3eca22924d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-11-15 17:57:29.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"45a33b8d909b44050538ed7172d692c35167d296","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 18:00:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7556c5f8f0fa4117bbba99d8ea3ce3fb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1594e044eb5e472a945150f06c1754bb : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c98472edf7e45b685d34a40cc5a3fd0 : SUCCESS in 7m 50s","accounts_in_message":[],"_revision_number":3},{"id":"7f0d9fb292f8bc0666d8196d43eab3855d25ba7d","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-11-23 08:38:47.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"383837b096ca5e7da2d6fd8b7e3fd4a5ccd1b2f6","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-11-23 09:55:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f023b52db4332cf6b913992f957413297ef9be7b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-25 02:10:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7475cc3371f6a4e120a2534301a556adb635dfb1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-25 03:29:01.000000000","message":"Topic set to policy-defaults-improvement","accounts_in_message":[],"_revision_number":3},{"id":"25e0a21c0169f17c70a57d2073e4a79ea29d27ba","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-30 04:04:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4bb18e3775d9184abdd3202491e6cd6e8cb65825","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-30 04:05:26.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"619c65d71cb321838a64164f4726d7a8269f3d8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-30 04:08:27.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"62ffd8856819e7f54cc93c083684faa90b873943","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 04:18:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc6d84d2e61e4a8bb015de2e8f373107\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00cc4a9960f243418c64b80efe1202d3 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e191df930cbd4104809fd668e2b54ffe : SUCCESS in 8m 41s","accounts_in_message":[],"_revision_number":5},{"id":"5108dbab14a2cb01eb799f4adfc9fbdef3337f95","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-11-30 08:24:02.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3d6b30226cf2b9d1a13c7f9f7a59ffd544db0e49","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-11-30 11:58:36.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3a28bcea8a001dfaf0dbd04db6ca10e40ecc17d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 11:59:06.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"dd5ab90ad8993076566fe2e82c2a322efd63c13c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 12:18:47.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1de70e1ff92b47798ceeee2f13caa755\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be0ea02d274e433baa1dd4f490349eab : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcb78ad56d334fef900a52fd064e00de : SUCCESS in 7m 28s","accounts_in_message":[],"_revision_number":5},{"id":"8cb34f815e2a2059d76a64c197d9b3e110d9eabb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 12:18:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"522242916f6bbbab173b9b610b951631e6890057","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 12:20:55.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/416b6a9b454d4ed88d4d4bb6a233bdd6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffe304f220d8416bb8bf4613091dcb90 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"bacd2c039345dd12f22b0474b8a0f6a691a8ed98","revisions":{"d2f10a02aab6028ca8648aff011c8490fc324fdd":{"kind":"REWORK","_number":1,"created":"2022-11-14 05:19:59.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/85/864385/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/85/864385/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/864385/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/864385/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/864385/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/85/864385/1"}}},"commit":{"parents":[{"commit":"4e345e55c2d695f7c271ac5119cfed761eb7d206","subject":"Switch to 2023.1 Python3 unit tests and generic template name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/4e345e55c2d695f7c271ac5119cfed761eb7d206"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-11-14 05:17:35.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-11-14 05:17:35.000000000","tz":-360},"subject":"Policy defaults improvement spec","message":"Policy defaults improvement spec\n\nThis spec is to improve the policies default\nas per the community wide goal\n- https://governance.openstack.org/tc/goals/selected/consistent-and-secure-rbac.html\n\nAPIImpact\n\nChange-Id: I12dce16a7f18bb065e1412d2f2d9bd91452d8672\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d2f10a02aab6028ca8648aff011c8490fc324fdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d2f10a02aab6028ca8648aff011c8490fc324fdd"}]},"branch":"refs/heads/master"},"98f361d8639400e710c42fefc8fc2ae5aa901b36":{"kind":"REWORK","_number":2,"created":"2022-11-15 16:34:22.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/85/864385/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/85/864385/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/864385/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/864385/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/864385/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/85/864385/2"}}},"commit":{"parents":[{"commit":"4e345e55c2d695f7c271ac5119cfed761eb7d206","subject":"Switch to 2023.1 Python3 unit tests and generic template name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/4e345e55c2d695f7c271ac5119cfed761eb7d206"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-11-14 05:17:35.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-11-15 16:34:12.000000000","tz":-360},"subject":"Policy defaults improvement spec","message":"Policy defaults improvement spec\n\nThis spec is to improve the policies default\nas per the community wide goal\n- https://governance.openstack.org/tc/goals/selected/consistent-and-secure-rbac.html\n\nAPIImpact\n\nChange-Id: I12dce16a7f18bb065e1412d2f2d9bd91452d8672\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/98f361d8639400e710c42fefc8fc2ae5aa901b36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/98f361d8639400e710c42fefc8fc2ae5aa901b36"}]},"branch":"refs/heads/master"},"7d9771cfce2d072b22026f2d83815f045816ad06":{"kind":"REWORK","_number":3,"created":"2022-11-15 17:51:12.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/85/864385/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/85/864385/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/864385/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/864385/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/864385/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/85/864385/3"}}},"commit":{"parents":[{"commit":"4e345e55c2d695f7c271ac5119cfed761eb7d206","subject":"Switch to 2023.1 Python3 unit tests and generic template name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/4e345e55c2d695f7c271ac5119cfed761eb7d206"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-11-14 05:17:35.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-11-15 17:51:04.000000000","tz":-360},"subject":"Policy defaults improvement spec","message":"Policy defaults improvement spec\n\nThis spec is to improve the policies default\nas per the community wide goal\n- https://governance.openstack.org/tc/goals/selected/consistent-and-secure-rbac.html\n\nAPIImpact\n\nChange-Id: I12dce16a7f18bb065e1412d2f2d9bd91452d8672\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/7d9771cfce2d072b22026f2d83815f045816ad06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/7d9771cfce2d072b22026f2d83815f045816ad06"}]},"branch":"refs/heads/master"},"ca8e791217ed30891e8174ae11551964352132a4":{"kind":"REWORK","_number":4,"created":"2022-11-30 04:05:26.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/85/864385/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/85/864385/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/864385/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/864385/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/864385/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/85/864385/4"}}},"commit":{"parents":[{"commit":"4e345e55c2d695f7c271ac5119cfed761eb7d206","subject":"Switch to 2023.1 Python3 unit tests and generic template name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/4e345e55c2d695f7c271ac5119cfed761eb7d206"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-11-14 05:17:35.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2022-11-30 04:05:26.000000000","tz":0},"subject":"Policy defaults improvement spec","message":"Policy defaults improvement spec\n\nThis spec is to improve the policies default\nas per the community wide goal\n- https://governance.openstack.org/tc/goals/selected/consistent-and-secure-rbac.html\n\nAPIImpact\n\nChange-Id: I12dce16a7f18bb065e1412d2f2d9bd91452d8672\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ca8e791217ed30891e8174ae11551964352132a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ca8e791217ed30891e8174ae11551964352132a4"}]},"branch":"refs/heads/master"},"bacd2c039345dd12f22b0474b8a0f6a691a8ed98":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-11-30 04:08:27.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/85/864385/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/85/864385/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/864385/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/864385/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/864385/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/85/864385/5"}}},"commit":{"parents":[{"commit":"4e345e55c2d695f7c271ac5119cfed761eb7d206","subject":"Switch to 2023.1 Python3 unit tests and generic template name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/4e345e55c2d695f7c271ac5119cfed761eb7d206"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-11-14 05:17:35.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2022-11-30 04:08:27.000000000","tz":0},"subject":"Policy defaults improvement spec","message":"Policy defaults improvement spec\n\nThis spec is to improve the policies default\nas per the community wide goal\n- https://governance.openstack.org/tc/goals/selected/consistent-and-secure-rbac.html\n\nAPIImpact\n\nPartial implement blueprint policy-defaults-improvement\n\nChange-Id: I12dce16a7f18bb065e1412d2f2d9bd91452d8672\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/bacd2c039345dd12f22b0474b8a0f6a691a8ed98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/bacd2c039345dd12f22b0474b8a0f6a691a8ed98"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
