)]}'
{"id":"openstack%2Fplacement~868354","triplet_id":"openstack%2Fplacement~master~Id49ad5a60f7f48f5718a7e102a946f452c45a41d","project":"openstack/placement","branch":"master","topic":"force-tox-v4","attention_set":{},"removed_from_attention_set":{"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2022-12-22 14:51:35.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id49ad5a60f7f48f5718a7e102a946f452c45a41d","subject":"DNM force latest tox","status":"ABANDONED","created":"2022-12-21 15:52:06.000000000","updated":"2022-12-22 14:51:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2bc07c9d3a98528b550eec39271761441af99791","_number":868354,"virtual_id_number":868354,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-12-21 17:44:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-21 17:44:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c1fff964ee5a894a8f4aa2974fa3eadaf8cedcbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-12-21 15:52:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"009106b18473ef340c68936437a69e4b7ed68384","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 17:44:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1cfa65e76c10436a960c3a211143c16f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ac79f8b152cd40608cca2cc75c34fb51 : FAILURE in 5m 00s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6eb5285cc731445d806650e9cfd0c667 : SUCCESS in 2m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f59b3bc34abf46b28ece4ee0e24fb21c : SUCCESS in 3m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/cb2271ce19e049adac8b649668683085 : SUCCESS in 1h 12m 43s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/fa1b46ca9ec24cbdb8ead7bdc57f47f1 : SUCCESS in 51m 44s (non-voting)\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/f84a90d30ae34e9b9ce38686792c8975 : SUCCESS in 1h 36m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c08a99e0992c4da38d2e7dbbfa3530d7 : FAILURE in 29m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6df850af637e4b7f9d75f984fb1c18c3 : FAILURE in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/885188526dc643b3a7213b65cf465c16 : FAILURE in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9e03e8af4e24e6a8d531ebf050015cb : FAILURE in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d5b679902c254477a225f22630fa79f1 : FAILURE in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1096b25fe1a4bda8b3cf4081e7e3810 : FAILURE in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81462ac1d1c046aa8ca8c1f22dc2e8f9 : SUCCESS in 5m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/566f3669d8f440be84332c1dc8165c29 : FAILURE in 4m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bcdc1e0ff34b482d889caff1de4bb293 : FAILURE in 4m 16s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ff3b9e172c554cebb933e71f4d547659 : SUCCESS in 16m 44s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/c138152634994ba787094707d3a65fe5 : SUCCESS in 21m 15s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/ad34c9ca88a540539c7b8af8f0f1de5e : SUCCESS in 10m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea26e6457adb411ca121688451433b0a : SUCCESS in 51m 55s","accounts_in_message":[],"_revision_number":1},{"id":"2bc07c9d3a98528b550eec39271761441af99791","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-12-22 14:51:35.000000000","message":"Abandoned\n\nfix is in https://review.opendev.org/c/openstack/placement/+/868418","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"99786a35e318b250df7553b1b6291bef861ffb5c","revisions":{"99786a35e318b250df7553b1b6291bef861ffb5c":{"kind":"REWORK","_number":1,"created":"2022-12-21 15:52:06.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/54/868354/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/54/868354/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/54/868354/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/54/868354/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/54/868354/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/54/868354/1"}}},"commit":{"parents":[{"commit":"bacd2c039345dd12f22b0474b8a0f6a691a8ed98","subject":"Policy defaults improvement spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/bacd2c039345dd12f22b0474b8a0f6a691a8ed98"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2022-12-21 15:50:43.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2022-12-21 15:52:02.000000000","tz":60},"subject":"DNM force latest tox","message":"DNM force latest tox\n\nThis change will test placement against tox v4 to check compatibility.\n\nChange-Id: Id49ad5a60f7f48f5718a7e102a946f452c45a41d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/99786a35e318b250df7553b1b6291bef861ffb5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/99786a35e318b250df7553b1b6291bef861ffb5c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
