)]}'
{"id":"openstack%2Fplacement~868727","triplet_id":"openstack%2Fplacement~stable%2Fzed~I1de24b7a98f149fce33d4e5eca82460eea72140b","project":"openstack/placement","branch":"stable/zed","topic":"tox4","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-01-05 18:53:06.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I1de24b7a98f149fce33d4e5eca82460eea72140b","subject":"Make tox.ini tox 4.0.0 compatible","status":"ABANDONED","created":"2022-12-27 20:14:08.000000000","updated":"2023-01-07 00:02:49.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7dded9ef6b37f07027877a7b8b026399968d87c6","_number":868727,"virtual_id_number":868727,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-12-27 21:24:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":-1,"date":"2023-01-04 11:24:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":-1,"date":"2023-01-05 18:53:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-27 20:14:08.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2022-12-27 21:24:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-04 11:24:39.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"abfc29e7c232f62fd31da8a01013e817bb26b429","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-27 20:14:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"400652bb4ebbf3f2ce6d256b93ef6e30d1eda646","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-27 20:25:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5798eca9408bc2d1e3b92581bd74cc657cc6b227","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-27 21:24:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e73b9dfaad14d2a8485a76e3e88e1ea\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e8f430f6dd3342db96c7fb5ce5f88b1c : SUCCESS in 54m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc5d43fb11044fffb2c4152963ece926 : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/840d72fc655047149fe3accbab175805 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/555bbcacf75f4164b06aeb8c10522f78 : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a119e4c06cba4c50911b15b6e1120b0b : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4e05164e5d2f4ee7923db0c34e932025 : RETRY_LIMIT in 1m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06c02e6ab9a845dea37258543657e737 : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4c58a0be2894bb6bbcb964462a951d1 : SUCCESS in 4m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b7c02f0303584480b26457cd52a9fd63 : SUCCESS in 4m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f4abad9275b54d29864fcc21b7c14d49 : SUCCESS in 4m 50s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a48ded8866314ccbb67662369a9a2a53 : SUCCESS in 18m 37s","accounts_in_message":[],"_revision_number":2},{"id":"2844bcd050079615359ac0ce673a376f1ce4eb62","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-01-04 11:24:39.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"36ef8eb3a55d2937fd5a6a25025dfa98242e835c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-05 18:53:06.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7dded9ef6b37f07027877a7b8b026399968d87c6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-07 00:02:49.000000000","message":"Abandoned\n\ntox is pinned \u003c4 for stable branches https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/867849","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9835a24f443e55211ea0f658fe801d9630149ba8","revisions":{"59b0e23bd9beb9aa12763dada14eea01d8b6163a":{"kind":"REWORK","_number":1,"created":"2022-12-27 20:14:08.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/27/868727/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/27/868727/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/27/868727/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/27/868727/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/27/868727/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/27/868727/1"}}},"commit":{"parents":[{"commit":"d7ced6bd2fc82caf458f20b5652888164b1bbb70","subject":"Update TOX_CONSTRAINTS_FILE for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d7ced6bd2fc82caf458f20b5652888164b1bbb70"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2022-12-22 14:46:04.000000000","tz":60},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-27 20:11:58.000000000","tz":-360},"subject":"Make tox.ini tox 4.0.0 compatible","message":"Make tox.ini tox 4.0.0 compatible\n\n* removed skipsdist\u003dTrue to make sure placement available in the virtual\n  env. This was at least needed for the doc generation\n\n* doc target uses requirements (e.g. six, oslo libs) from requirements.txt\n  not just from doc/requirements.txt\n\n* passenv fixed as space separated list is not allowed any more\n\n* improved the doc generation command line to print all the errors\n  instead of stopping at the first one\n\nExtra updates in backport:\nfunctional-py310 tox env was added in master commit but that is not needed\nfor stable branches so removing that in backports.\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/866943\nChange-Id: I1de24b7a98f149fce33d4e5eca82460eea72140b\n(cherry picked from commit 8bb18ecb25ad1076aa594408c5ca360b9cf33c9e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/59b0e23bd9beb9aa12763dada14eea01d8b6163a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/59b0e23bd9beb9aa12763dada14eea01d8b6163a"}]},"branch":"refs/heads/stable/zed"},"9835a24f443e55211ea0f658fe801d9630149ba8":{"kind":"REWORK","_number":2,"created":"2022-12-27 20:25:22.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/27/868727/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/27/868727/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/27/868727/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/27/868727/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/27/868727/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/27/868727/2"}}},"commit":{"parents":[{"commit":"d7ced6bd2fc82caf458f20b5652888164b1bbb70","subject":"Update TOX_CONSTRAINTS_FILE for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d7ced6bd2fc82caf458f20b5652888164b1bbb70"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2022-12-22 14:46:04.000000000","tz":60},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-27 20:24:49.000000000","tz":-360},"subject":"Make tox.ini tox 4.0.0 compatible","message":"Make tox.ini tox 4.0.0 compatible\n\n* removed skipsdist\u003dTrue to make sure placement available in the virtual\n  env. This was at least needed for the doc generation\n\n* doc target uses requirements (e.g. six, oslo libs) from requirements.txt\n  not just from doc/requirements.txt\n\n* passenv fixed as space separated list is not allowed any more\n\n* improved the doc generation command line to print all the errors\n  instead of stopping at the first one\n\nExtra updates in backport:\n* Updated constraints to stable\n* functional-py310 tox env was added in master commit but that is not needed\n  for stable branches so removing that in backports.\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/866943\nChange-Id: I1de24b7a98f149fce33d4e5eca82460eea72140b\n(cherry picked from commit 8bb18ecb25ad1076aa594408c5ca360b9cf33c9e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9835a24f443e55211ea0f658fe801d9630149ba8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9835a24f443e55211ea0f658fe801d9630149ba8"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
