)]}'
{"id":"openstack%2Fplacement~868735","triplet_id":"openstack%2Fplacement~stable%2Fyoga~I1de24b7a98f149fce33d4e5eca82460eea72140b","project":"openstack/placement","branch":"stable/yoga","topic":"tox4","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-01-05 18:52:59.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I1de24b7a98f149fce33d4e5eca82460eea72140b","subject":"Make tox.ini tox 4.0.0 compatible","status":"ABANDONED","created":"2022-12-28 05:22:50.000000000","updated":"2023-01-07 00:03:01.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1ee41dc2c2373b90140ca0bdb91b76e437f755d0","_number":868735,"virtual_id_number":868735,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-12-30 00:26:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":-1,"date":"2023-01-04 11:24:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":-1,"date":"2023-01-05 18:52:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-28 05:22:50.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2022-12-28 06:13:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-04 11:24:54.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"42249074ae603d7d54524c1e283efdbbf1fce32b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-28 05:22:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f305b5dc8528d2ab5241c1fa7eed0259b3667aa9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-28 05:24:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d7c5422abf0358e4bf0d0dccab079f43cb1740a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 06:13:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1fc023ba55541ffb8e2f73a0d5b616e\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/688b7d2deb5a40d89e839224e4798ec5 : SUCCESS in 48m 09s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/21f5cae7e0ea48528613f5157c81005c : SUCCESS in 44m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad21c0dafe744e9ca5fa90d10232185a : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44def8b06286484f9057875e9965c086 : FAILURE in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25225dbffa004c8692c7fffd50d62e33 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a74d69780e0475695028b4a07f95aab : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a76525f9d7a4a83960e483e495d4f5b : FAILURE in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd78e649f1694f44be8e755212152896 : SUCCESS in 4m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/282e909a09d645a9bef7073e16f37429 : SUCCESS in 4m 45s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/9b74451ca00240749591d75fe4e85138 : SUCCESS in 4m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0498ef5e261e4a9fb54bdc6dc59a3006 : SUCCESS in 5m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c059c870d3464dcfa4a5658d2ad9a883 : SUCCESS in 6m 03s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1ba7c9be959a42829333139cb0df76f2 : FAILURE in 6m 15s","accounts_in_message":[],"_revision_number":2},{"id":"84fcf18c54728ed6dca1508e399660223fe8a579","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-28 19:36:20.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"263846d9c2fd7744c48705f8f60858d81cd447ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 20:29:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/50eea50a25754190996471c0a15c5e61\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/47f8a052afaa487ba99f570382f15e97 : SUCCESS in 51m 57s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bfd3d82907dc4b3eaa3140f3cc1f780a : SUCCESS in 43m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd9d7233aea9479f979232dd42304809 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecb2c82d7e9e4f36af052ccb7744a5c7 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3354f3bae8a345ed90203256028004bf : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae7821c3ef0340b9ab0621f32c946dfd : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a974fb3583744a71824eb016a783c5d3 : FAILURE in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fce3c62b87fd41e0be022d5f967c87a5 : SUCCESS in 4m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e2b76ee78df046a88f91280fa26335b7 : SUCCESS in 4m 59s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/8e2ede154ad0447892af06c1084b8629 : SUCCESS in 5m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/346db00f27244d5d819736fcae075ba9 : SUCCESS in 5m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d83129d1692248d1ac35af8468effce1 : SUCCESS in 5m 39s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ace11aa55cd1402684af096bd0043f91 : SUCCESS in 14m 36s","accounts_in_message":[],"_revision_number":3},{"id":"6915a1ec222db40b3e2826c88f76337fad85f139","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-29 23:10:24.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"28eaa83e6cf697ee0d7ce442b20d4121bf35b390","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-30 00:26:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e35a0eec7ea14f969326c7e881dd2390\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/07a2a8fef61844c093cf1f15e04ab8b5 : SUCCESS in 1h 10m 37s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8ec7e4af31c4cbc88aadb2d39c4b9c6 : SUCCESS in 46m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/835e6f47bcc24f38ab72a926de6bcd25 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75c6b3bb096046288b43050253d5f78f : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/256e7179ecab44159c9f33c3a250c6ca : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b09e93c427243c6b20d60dfc228ebd1 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6352f11ae52a42eb93e17661eae41a0c : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37451df5afd84a118c695872525505bc : SUCCESS in 7m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/97b2e525e1d54b939e8e4a43506e955d : SUCCESS in 5m 56s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/43ad152d82f746468a325f8b9afbcd1b : SUCCESS in 5m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2c9cf258188243938866b43e825b27f1 : SUCCESS in 6m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bfec0030374f47b1a0e75c61be72154b : SUCCESS in 8m 31s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/606b4718cd6545e4a1f4e0f57b394d81 : SUCCESS in 18m 34s","accounts_in_message":[],"_revision_number":4},{"id":"dce3e8b589df7e49a90ddee8373ecbd38bc9dd76","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-01-04 11:24:54.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d09aa00b7e2ac417270523d042e72924a1f73dd7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-05 18:52:59.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1ee41dc2c2373b90140ca0bdb91b76e437f755d0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-07 00:03:01.000000000","message":"Abandoned\n\ntox is pinned \u003c4 for stable branches https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/867849","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"dd64f59d44b055c15bc983b21955f256dcad689d","revisions":{"6fd45f6880dd341f2cec1095876539c10ab23ea2":{"kind":"REWORK","_number":1,"created":"2022-12-28 05:22:50.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/35/868735/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/35/868735/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/35/868735/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/35/868735/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/35/868735/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/35/868735/1"}}},"commit":{"parents":[{"commit":"4d3df47ee3e394e3178d58c15306620809ad2806","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/4d3df47ee3e394e3178d58c15306620809ad2806"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2022-12-22 14:46:04.000000000","tz":60},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-28 05:22:29.000000000","tz":-360},"subject":"Make tox.ini tox 4.0.0 compatible","message":"Make tox.ini tox 4.0.0 compatible\n\n* removed skipsdist\u003dTrue to make sure placement available in the virtual\n  env. This was at least needed for the doc generation\n\n* doc target uses requirements (e.g. six, oslo libs) from requirements.txt\n  not just from doc/requirements.txt\n\n* passenv fixed as space separated list is not allowed any more\n\n* improved the doc generation command line to print all the errors\n  instead of stopping at the first one\n\nExtra updates in backport:\n* Updated constraints to stable\n* functional-py310 tox env was added in master commit but that is not needed\n  for stable branches so removing that in backports.\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/866943\nChange-Id: I1de24b7a98f149fce33d4e5eca82460eea72140b\n(cherry picked from commit 8bb18ecb25ad1076aa594408c5ca360b9cf33c9e)\n(cherry picked from commit 9835a24f443e55211ea0f658fe801d9630149ba8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/6fd45f6880dd341f2cec1095876539c10ab23ea2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/6fd45f6880dd341f2cec1095876539c10ab23ea2"}]},"branch":"refs/heads/stable/yoga"},"240e4b775ec8973c079416237b3f2ac1bccd5313":{"kind":"REWORK","_number":2,"created":"2022-12-28 05:24:18.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/35/868735/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/35/868735/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/35/868735/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/35/868735/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/35/868735/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/35/868735/2"}}},"commit":{"parents":[{"commit":"4d3df47ee3e394e3178d58c15306620809ad2806","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/4d3df47ee3e394e3178d58c15306620809ad2806"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2022-12-22 14:46:04.000000000","tz":60},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-28 05:24:11.000000000","tz":-360},"subject":"Make tox.ini tox 4.0.0 compatible","message":"Make tox.ini tox 4.0.0 compatible\n\n* removed skipsdist\u003dTrue to make sure placement available in the virtual\n  env. This was at least needed for the doc generation\n\n* doc target uses requirements (e.g. six, oslo libs) from requirements.txt\n  not just from doc/requirements.txt\n\n* passenv fixed as space separated list is not allowed any more\n\n* improved the doc generation command line to print all the errors\n  instead of stopping at the first one\n\nExtra updates in backport:\n* Updated constraints to stable\n* functional-py310 tox env was added in master commit but that is not needed\n  for stable branches so removing that in backports.\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/866943\nChange-Id: I1de24b7a98f149fce33d4e5eca82460eea72140b\n(cherry picked from commit 8bb18ecb25ad1076aa594408c5ca360b9cf33c9e)\n(cherry picked from commit 9835a24f443e55211ea0f658fe801d9630149ba8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/240e4b775ec8973c079416237b3f2ac1bccd5313"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/240e4b775ec8973c079416237b3f2ac1bccd5313"}]},"branch":"refs/heads/stable/yoga"},"e5e030e248a87089d447658111fb98e2d5191370":{"kind":"REWORK","_number":3,"created":"2022-12-28 19:36:20.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/35/868735/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/35/868735/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/35/868735/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/35/868735/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/35/868735/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/35/868735/3"}}},"commit":{"parents":[{"commit":"4d3df47ee3e394e3178d58c15306620809ad2806","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/4d3df47ee3e394e3178d58c15306620809ad2806"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2022-12-22 14:46:04.000000000","tz":60},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-28 19:36:20.000000000","tz":0},"subject":"Make tox.ini tox 4.0.0 compatible","message":"Make tox.ini tox 4.0.0 compatible\n\n* removed skipsdist\u003dTrue to make sure placement available in the virtual\n  env. This was at least needed for the doc generation\n\n* doc target uses requirements (e.g. six, oslo libs) from requirements.txt\n  not just from doc/requirements.txt\n\n* passenv fixed as space separated list is not allowed any more\n\n* improved the doc generation command line to print all the errors\n  instead of stopping at the first one\n\nExtra updates in backport:\n* Updated constraints to stable\n* functional-py310 tox env was added in master commit but that is not needed\n  for stable branches so removing that in backports.\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/866943\nDepends-On: https://review.opendev.org/c/openstack/nova/+/868728/3\n\nChange-Id: I1de24b7a98f149fce33d4e5eca82460eea72140b\n(cherry picked from commit 8bb18ecb25ad1076aa594408c5ca360b9cf33c9e)\n(cherry picked from commit 9835a24f443e55211ea0f658fe801d9630149ba8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e5e030e248a87089d447658111fb98e2d5191370"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e5e030e248a87089d447658111fb98e2d5191370"}]},"branch":"refs/heads/stable/yoga"},"dd64f59d44b055c15bc983b21955f256dcad689d":{"kind":"REWORK","_number":4,"created":"2022-12-29 23:10:24.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/35/868735/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/35/868735/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/35/868735/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/35/868735/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/35/868735/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/35/868735/4"}}},"commit":{"parents":[{"commit":"4d3df47ee3e394e3178d58c15306620809ad2806","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/4d3df47ee3e394e3178d58c15306620809ad2806"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2022-12-22 14:46:04.000000000","tz":60},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-29 23:10:17.000000000","tz":0},"subject":"Make tox.ini tox 4.0.0 compatible","message":"Make tox.ini tox 4.0.0 compatible\n\n* removed skipsdist\u003dTrue to make sure placement available in the virtual\n  env. This was at least needed for the doc generation\n\n* doc target uses requirements (e.g. six, oslo libs) from requirements.txt\n  not just from doc/requirements.txt\n\n* passenv fixed as space separated list is not allowed any more\n\n* improved the doc generation command line to print all the errors\n  instead of stopping at the first one\n\nExtra updates in backport:\n* Updated constraints to stable\n* functional-py310 tox env was added in master commit but that is not needed\n  for stable branches so removing that in backports.\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/866943\nDepends-On: https://review.opendev.org/c/openstack/nova/+/868728/3\n\nChange-Id: I1de24b7a98f149fce33d4e5eca82460eea72140b\n(cherry picked from commit 8bb18ecb25ad1076aa594408c5ca360b9cf33c9e)\n(cherry picked from commit 9835a24f443e55211ea0f658fe801d9630149ba8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/dd64f59d44b055c15bc983b21955f256dcad689d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/dd64f59d44b055c15bc983b21955f256dcad689d"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
