)]}'
{"id":"openstack%2Fplacement~868736","triplet_id":"openstack%2Fplacement~stable%2Fxena~I1de24b7a98f149fce33d4e5eca82460eea72140b","project":"openstack/placement","branch":"stable/xena","topic":"tox4","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-01-05 18:52:51.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I1de24b7a98f149fce33d4e5eca82460eea72140b","subject":"Make tox.ini tox 4.0.0 compatible","status":"ABANDONED","created":"2022-12-28 05:26:28.000000000","updated":"2023-01-07 00:03:09.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"67b8996f6587b1079d8bb3b608cbaa3a7ddc7c87","_number":868736,"virtual_id_number":868736,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-12-30 18:33:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":-1,"date":"2023-01-04 11:25:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":-1,"date":"2023-01-05 18:52:51.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-28 05:26:28.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2022-12-28 06:09:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-04 11:25:02.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"3b7838ad9af11786a5ce7fde015ed7da67c7ebe4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-28 05:26:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"068a06a2be43bb9aea8bb699a73970bcd52294de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 06:09:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cefd30087ba546ee91172bf6f181d840\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df07b96e9ce34e6498257ae860a71f6f : SUCCESS in 41m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f45128214544ae083494741810498ce : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7824c9c2d0af41a18bf9f7dd26e83c04 : FAILURE in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a6473d3708c49f9bef1b281912ae36f : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a170ffd295154692a61bcfe05c9a7db5 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/382ef2b5b247401a85e502241f6bd084 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4835ade9b98844138dfbdc360a8c611f : FAILURE in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a87e3e4b145416a89bd7e62533037b4 : SUCCESS in 6m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6e98bf68fa1a40f0a6508f8e6cde9212 : SUCCESS in 4m 27s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/166a17f08b224b8db2378ca166bc749e : SUCCESS in 7m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9cd44dd8ee224efc8ca74a37b0463dcb : SUCCESS in 5m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3478b9fde6e74a3bb7fae550801c478b : SUCCESS in 4m 45s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/63a54df821ac49e1bb4ac131d4300ca5 : FAILURE in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"1a3207e19fae1016ea4e5e25be98af96374bada6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-28 20:41:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1acef95e25cebbd6ec2afd256f240f8b4dfa5e77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 21:41:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/86d500eaad364714b57b4af953a2ec6f\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bba4e2e7e4f34c2b91466fa39cfc1ba5 : SUCCESS in 54m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95c7e504510d4e8fa13059acf2506f81 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45c77b4f74f04f0badf6585a0fe9ee5d : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e825c02d177049d2bef27a50272aed00 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f747691a94f144c38a8d9cf14c9425f5 : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d7d67c073bf486d9b5f5eac6f729c08 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1018b5b017584da5bb0951272644d497 : FAILURE in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d497f859b17c40b4b6e83f88e9712b5a : SUCCESS in 5m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8b8a0a80608c4e9ab849446d67804f3b : SUCCESS in 4m 10s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/2cd8e30a7471462d911624bee28c6e00 : SUCCESS in 7m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7e8c624377f84497a15c7d7c490c1ff8 : SUCCESS in 5m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8efc185a7ca3439f89ce5fc80480d311 : SUCCESS in 6m 26s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9f39b05c983c40c7bfcffa3bacfe919d : FAILURE in 3m 35s","accounts_in_message":[],"_revision_number":2},{"id":"632fc3c99f2eca6fb19fb221ce5eb941e5cb6726","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-29 23:36:39.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"d24494640e470a271ee19b2c35e99490e0c862e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-30 00:16:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7cdc3675ad4b4152aa879406359d5ece\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/76ef1828b4e741af9684e41758bf722a : SUCCESS in 34m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b639eb9a60e4d18ae5cc4daca3e5988 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91524a372d6f429baf79569a1232c39a : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b033a47fadbb47a6b57a7663fef9c279 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78326b86645c475ca791697e39277e93 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10af6f914eb94ce395aacae122649ca8 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8c14dc6dcb143f8a9f2f026563c0d99 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd6d3e3f4f854793bf171848543b243e : SUCCESS in 4m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6cc15be0507e4b43bc4114d2f973fd79 : SUCCESS in 5m 43s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/03165d93e71d49c397d6f56312934419 : SUCCESS in 5m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ddf7ed324a2d456486f864621c721e76 : SUCCESS in 5m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1453b6c460c945928be7a5d4cf6d9b42 : SUCCESS in 5m 03s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f47196063c994e2287e1b982b3841ad7 : FAILURE in 4m 15s","accounts_in_message":[],"_revision_number":3},{"id":"1f5b2eb30d190e2d3775f1d8356309240f20dda7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-30 17:30:44.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"5c91d425b1e033ed095aa84191bbda515b6a3f39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-30 18:33:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/439788d3aae940b5ba89bbae9ae0e721\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c53aab842524b16a786a20120e0cc49 : SUCCESS in 46m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79ccb972d7d348a88e851ccacfbd12c2 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9079f0d6d3b4371be74b944e707aac5 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5eaa1e710f54b289ff758204429e715 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ebf9379c97e14b34881d133eba63239c : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3be4cef751240048c29195211f0e274 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94fbd6ab367047b2a127e81f78097eae : SUCCESS in 8m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03faa49dc13d448f943a8841b91f11db : SUCCESS in 4m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fab25aaefceb4625aed4f042bd2a2d2d : SUCCESS in 5m 12s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/8ac04de4f849438eaa55a27300e83cfe : SUCCESS in 4m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/598fa59739634128a04b5bacd7ec03c8 : SUCCESS in 5m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dd28d26d35fb47cb9b96d9f7d1611bca : SUCCESS in 6m 31s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d9d31a5a357c4a098eb285bf281c78a2 : SUCCESS in 13m 40s","accounts_in_message":[],"_revision_number":4},{"id":"ba5a3e496981ac28522171c2533e3961a2e8a19f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-01-04 11:25:02.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bb712cbd0654c54a5f1d7622782f82338fa1f81f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-05 18:52:51.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"67b8996f6587b1079d8bb3b608cbaa3a7ddc7c87","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-07 00:03:09.000000000","message":"Abandoned\n\ntox is pinned \u003c4 for stable branches https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/867849","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9c17edbb014693122041b41d72a064f9383b681a","revisions":{"52ed8d8e013e83bf9c38c58d029e83264e5f8c95":{"kind":"REWORK","_number":1,"created":"2022-12-28 05:26:28.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/36/868736/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/36/868736/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/36/868736/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/36/868736/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/36/868736/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/36/868736/1"}}},"commit":{"parents":[{"commit":"f96984abd48f4d938d0fff0bab9b12ed933e320e","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f96984abd48f4d938d0fff0bab9b12ed933e320e"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2022-12-22 14:46:04.000000000","tz":60},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-28 05:26:12.000000000","tz":-360},"subject":"Make tox.ini tox 4.0.0 compatible","message":"Make tox.ini tox 4.0.0 compatible\n\n* removed skipsdist\u003dTrue to make sure placement available in the virtual\n  env. This was at least needed for the doc generation\n\n* doc target uses requirements (e.g. six, oslo libs) from requirements.txt\n  not just from doc/requirements.txt\n\n* passenv fixed as space separated list is not allowed any more\n\n* improved the doc generation command line to print all the errors\n  instead of stopping at the first one\n\nExtra updates in backport:\n* Updated constraints to stable\n* functional-py310 tox env was added in master commit but that is not needed\n  for stable branches so removing that in backports.\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/866943\nChange-Id: I1de24b7a98f149fce33d4e5eca82460eea72140b\n(cherry picked from commit 8bb18ecb25ad1076aa594408c5ca360b9cf33c9e)\n(cherry picked from commit 9835a24f443e55211ea0f658fe801d9630149ba8)\n(cherry picked from commit 240e4b775ec8973c079416237b3f2ac1bccd5313)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/52ed8d8e013e83bf9c38c58d029e83264e5f8c95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/52ed8d8e013e83bf9c38c58d029e83264e5f8c95"}]},"branch":"refs/heads/stable/xena"},"9b2e626241a3e17695641bd5a6fd84e0cb01ef3c":{"kind":"REWORK","_number":2,"created":"2022-12-28 20:41:42.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/36/868736/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/36/868736/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/36/868736/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/36/868736/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/36/868736/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/36/868736/2"}}},"commit":{"parents":[{"commit":"f96984abd48f4d938d0fff0bab9b12ed933e320e","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f96984abd48f4d938d0fff0bab9b12ed933e320e"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2022-12-22 14:46:04.000000000","tz":60},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-28 20:41:29.000000000","tz":-360},"subject":"Make tox.ini tox 4.0.0 compatible","message":"Make tox.ini tox 4.0.0 compatible\n\n* removed skipsdist\u003dTrue to make sure placement available in the virtual\n  env. This was at least needed for the doc generation\n\n* doc target uses requirements (e.g. six, oslo libs) from requirements.txt\n  not just from doc/requirements.txt\n\n* passenv fixed as space separated list is not allowed any more\n\n* improved the doc generation command line to print all the errors\n  instead of stopping at the first one\n\nExtra updates in backport:\n* Updated constraints to stable\n* functional-py310 tox env was added in master commit but that is not needed\n  for stable branches so removing that in backports.\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/866943\nDepends-On: https://review.opendev.org/c/openstack/nova/+/868728/3\n\nChange-Id: I1de24b7a98f149fce33d4e5eca82460eea72140b\n(cherry picked from commit 8bb18ecb25ad1076aa594408c5ca360b9cf33c9e)\n(cherry picked from commit 9835a24f443e55211ea0f658fe801d9630149ba8)\n(cherry picked from commit e5e030e248a87089d447658111fb98e2d5191370)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9b2e626241a3e17695641bd5a6fd84e0cb01ef3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9b2e626241a3e17695641bd5a6fd84e0cb01ef3c"}]},"branch":"refs/heads/stable/xena"},"2ad14b8201a88f49627e44d699b247caff1fbab4":{"kind":"REWORK","_number":3,"created":"2022-12-29 23:36:39.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/36/868736/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/36/868736/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/36/868736/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/36/868736/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/36/868736/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/36/868736/3"}}},"commit":{"parents":[{"commit":"f96984abd48f4d938d0fff0bab9b12ed933e320e","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f96984abd48f4d938d0fff0bab9b12ed933e320e"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2022-12-22 14:46:04.000000000","tz":60},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-29 23:36:39.000000000","tz":0},"subject":"Make tox.ini tox 4.0.0 compatible","message":"Make tox.ini tox 4.0.0 compatible\n\n* removed skipsdist\u003dTrue to make sure placement available in the virtual\n  env. This was at least needed for the doc generation\n\n* doc target uses requirements (e.g. six, oslo libs) from requirements.txt\n  not just from doc/requirements.txt\n\n* passenv fixed as space separated list is not allowed any more\n\n* improved the doc generation command line to print all the errors\n  instead of stopping at the first one\n\nExtra updates in backport:\n* Updated constraints to stable\n* functional-py310 tox env was added in master commit but that is not needed\n  for stable branches so removing that in backports.\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/866943\nDepends-On: https://review.opendev.org/c/openstack/nova/+/868728/3\n\nChange-Id: I1de24b7a98f149fce33d4e5eca82460eea72140b\n(cherry picked from commit 8bb18ecb25ad1076aa594408c5ca360b9cf33c9e)\n(cherry picked from commit 9835a24f443e55211ea0f658fe801d9630149ba8)\n(cherry picked from commit e5e030e248a87089d447658111fb98e2d5191370)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/2ad14b8201a88f49627e44d699b247caff1fbab4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/2ad14b8201a88f49627e44d699b247caff1fbab4"}]},"branch":"refs/heads/stable/xena"},"9c17edbb014693122041b41d72a064f9383b681a":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-12-30 17:30:44.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/36/868736/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/36/868736/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/36/868736/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/36/868736/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/36/868736/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/36/868736/4"}}},"commit":{"parents":[{"commit":"f96984abd48f4d938d0fff0bab9b12ed933e320e","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f96984abd48f4d938d0fff0bab9b12ed933e320e"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2022-12-22 14:46:04.000000000","tz":60},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-30 17:30:44.000000000","tz":0},"subject":"Make tox.ini tox 4.0.0 compatible","message":"Make tox.ini tox 4.0.0 compatible\n\n* removed skipsdist\u003dTrue to make sure placement available in the virtual\n  env. This was at least needed for the doc generation\n\n* doc target uses requirements (e.g. six, oslo libs) from requirements.txt\n  not just from doc/requirements.txt\n\n* passenv fixed as space separated list is not allowed any more\n\n* improved the doc generation command line to print all the errors\n  instead of stopping at the first one\n\nExtra updates in backport:\n* Updated constraints to stable\n* functional-py310 tox env was added in master commit but that is not needed\n  for stable branches so removing that in backports.\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/866943\nDepends-On: https://review.opendev.org/c/openstack/nova/+/868790\n\nChange-Id: I1de24b7a98f149fce33d4e5eca82460eea72140b\n(cherry picked from commit 8bb18ecb25ad1076aa594408c5ca360b9cf33c9e)\n(cherry picked from commit 9835a24f443e55211ea0f658fe801d9630149ba8)\n(cherry picked from commit e5e030e248a87089d447658111fb98e2d5191370)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9c17edbb014693122041b41d72a064f9383b681a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9c17edbb014693122041b41d72a064f9383b681a"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
