)]}'
{"id":"openstack%2Fplacement~868737","triplet_id":"openstack%2Fplacement~stable%2Fwallaby~I1de24b7a98f149fce33d4e5eca82460eea72140b","project":"openstack/placement","branch":"stable/wallaby","topic":"tox4","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-01-05 18:52:42.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I1de24b7a98f149fce33d4e5eca82460eea72140b","subject":"Make tox.ini tox 4.0.0 compatible","status":"ABANDONED","created":"2022-12-28 05:27:46.000000000","updated":"2023-01-07 00:02:40.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dec1ac6c4156fc35fd70be2fb039cb52703af348","_number":868737,"virtual_id_number":868737,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-12-30 18:40:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":-1,"date":"2023-01-04 11:25:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":-1,"date":"2023-01-05 18:52:42.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-28 05:27:46.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2022-12-28 06:23:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-04 11:25:10.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"04b0a3867632ccf302ca22b6be793df546db4123","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-28 05:27:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1592d6f5cbac4d64a36bfffa41395c05449f6c1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 06:23:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cbd9807e76d94ca08f43e81cba33d7cc\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71f2da5492d443baa86d11e98ceae00b : SUCCESS in 55m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8b569f41e7a4cf9a238814f9a006a34 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f382d2e7ea34b6c94ce50a8c7c2b044 : FAILURE in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d263afc26cdd4f24a684e93c49f59dc5 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/386a15fca8f1440a8e60e9cc721d4d46 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ff0f0faaf3c4763a772fa6bbf45a4c3 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a66114e8c894929a036708b49bfd81a : FAILURE in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0edc967cacb40bdaea64fbe632dd984 : SUCCESS in 4m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e9d37f2ee96c49e2b1924db7338cc2cc : SUCCESS in 4m 37s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/f25afa8dd5d24ba38b53f5c7c5c1b096 : SUCCESS in 5m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ca417494912547408b6023c1e88f6c48 : SUCCESS in 5m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e6f22db8ab45482c9a0966bd8c0dc0c4 : SUCCESS in 4m 51s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4d49a4886b374e97847a7ae1f2cc689a : FAILURE in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"55d48cb2c8150c2a06ad35f52b34bb95dcb64b4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-29 23:37:06.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"5772eec1a4cf9c065958fe7f4e98f61b5e63d501","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-30 00:47:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9991d2327ce64e6b9aae27809eb83c16\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0751d70ea58c4a078970124a0ec320f3 : SUCCESS in 1h 09m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3cb61c1cf750463f942ddf710264dff3 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76d6faadbeb742bab6fac15a1dd14242 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f5455618ca144e5b57b093a755c0278 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d042ffa664ba4ddea61b3e7ea9c3c91d : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75999fe23a454a42a5705d3f88ce3fd1 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f305f94baa7c4d2193333d004bbcfcdf : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23f6db6bb43540ad859e92bd1e2cdde8 : SUCCESS in 4m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/10af640de1774a04969a947814cc531f : SUCCESS in 5m 16s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/a6dc22f2752f48f7b165a710f76c4125 : SUCCESS in 5m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9485b3c5b9a4444c9277713582c2bd09 : SUCCESS in 5m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/eadd9f4fb1534ca3b0b31477b667dc17 : SUCCESS in 5m 19s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c227704dc4214394be5732a4279f778d : FAILURE in 3m 46s","accounts_in_message":[],"_revision_number":2},{"id":"27b35c501e166b5894a18fcb8f89de33d7d93fbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-30 17:31:18.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"f16c7eb1b942fe00c8bea5cac6763d9b63893959","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-30 18:40:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09097dc61eb6489bb63d965e0ec7416c\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a17bdb4016a4cfaac48073453d47f07 : SUCCESS in 1h 02m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec253c4c5a4649d89537a82a8364204e : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e7cf964cf104c1d94fb8e4568f8bd3f : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe182dc6f80e46d7a3b1c85a7038bf73 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/243d408cdb7c42e995bded7684461709 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/085fe806a11545408ecd2b08be9e87ec : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f89bec7349e34bc0bb55c0a09537bb38 : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25bd57749bf644b9ab7f1da01f873088 : SUCCESS in 5m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eefef2e0109845a58f5a6dec7650d95d : SUCCESS in 5m 25s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/a363e2185a6c4eb2bd9e23074d807ade : SUCCESS in 6m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e6cebd59393a42078d2b111cae8e8470 : SUCCESS in 5m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8be7449004924c6ab25ce41699219456 : SUCCESS in 5m 29s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7958a7a03d724e8288ab741734d2830e : SUCCESS in 14m 26s","accounts_in_message":[],"_revision_number":3},{"id":"8d74127458cde92c6b7c96aa0585d3a97db9d9c3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-01-04 11:25:10.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6d401173311e938bd524694a22da0c3426fe94d3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-05 18:52:42.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dec1ac6c4156fc35fd70be2fb039cb52703af348","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-07 00:02:40.000000000","message":"Abandoned\n\ntox is pinned \u003c4 for stable branches https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/867849","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6a04af2e76c8c8b8eab0ea084838aa2fcd599a11","revisions":{"7f68f89c615fbf236c59067d66d65baf33d10482":{"kind":"REWORK","_number":1,"created":"2022-12-28 05:27:46.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/37/868737/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/37/868737/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/37/868737/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/37/868737/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/37/868737/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/37/868737/1"}}},"commit":{"parents":[{"commit":"716b947c479e36ce2e9ec6d9e0bbab882f391892","subject":"Merge \"Add periodic-stable-jobs template\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/716b947c479e36ce2e9ec6d9e0bbab882f391892"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2022-12-22 14:46:04.000000000","tz":60},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-28 05:27:33.000000000","tz":-360},"subject":"Make tox.ini tox 4.0.0 compatible","message":"Make tox.ini tox 4.0.0 compatible\n\n* removed skipsdist\u003dTrue to make sure placement available in the virtual\n  env. This was at least needed for the doc generation\n\n* doc target uses requirements (e.g. six, oslo libs) from requirements.txt\n  not just from doc/requirements.txt\n\n* passenv fixed as space separated list is not allowed any more\n\n* improved the doc generation command line to print all the errors\n  instead of stopping at the first one\n\nExtra updates in backport:\n* Updated constraints to stable\n* functional-py310 tox env was added in master commit but that is not needed\n  for stable branches so removing that in backports.\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/866943\nChange-Id: I1de24b7a98f149fce33d4e5eca82460eea72140b\n(cherry picked from commit 8bb18ecb25ad1076aa594408c5ca360b9cf33c9e)\n(cherry picked from commit 9835a24f443e55211ea0f658fe801d9630149ba8)\n(cherry picked from commit 240e4b775ec8973c079416237b3f2ac1bccd5313)\n(cherry picked from commit 52ed8d8e013e83bf9c38c58d029e83264e5f8c95)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/7f68f89c615fbf236c59067d66d65baf33d10482"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/7f68f89c615fbf236c59067d66d65baf33d10482"}]},"branch":"refs/heads/stable/wallaby"},"41e413089284f78db1921c8ec1ec1e7dc4c2a765":{"kind":"REWORK","_number":2,"created":"2022-12-29 23:37:06.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/37/868737/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/37/868737/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/37/868737/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/37/868737/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/37/868737/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/37/868737/2"}}},"commit":{"parents":[{"commit":"716b947c479e36ce2e9ec6d9e0bbab882f391892","subject":"Merge \"Add periodic-stable-jobs template\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/716b947c479e36ce2e9ec6d9e0bbab882f391892"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2022-12-22 14:46:04.000000000","tz":60},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-29 23:37:05.000000000","tz":0},"subject":"Make tox.ini tox 4.0.0 compatible","message":"Make tox.ini tox 4.0.0 compatible\n\n* removed skipsdist\u003dTrue to make sure placement available in the virtual\n  env. This was at least needed for the doc generation\n\n* doc target uses requirements (e.g. six, oslo libs) from requirements.txt\n  not just from doc/requirements.txt\n\n* passenv fixed as space separated list is not allowed any more\n\n* improved the doc generation command line to print all the errors\n  instead of stopping at the first one\n\nExtra updates in backport:\n* Updated constraints to stable\n* functional-py310 tox env was added in master commit but that is not needed\n  for stable branches so removing that in backports.\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/866943\nChange-Id: I1de24b7a98f149fce33d4e5eca82460eea72140b\n(cherry picked from commit 8bb18ecb25ad1076aa594408c5ca360b9cf33c9e)\n(cherry picked from commit 9835a24f443e55211ea0f658fe801d9630149ba8)\n(cherry picked from commit 240e4b775ec8973c079416237b3f2ac1bccd5313)\n(cherry picked from commit 52ed8d8e013e83bf9c38c58d029e83264e5f8c95)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/41e413089284f78db1921c8ec1ec1e7dc4c2a765"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/41e413089284f78db1921c8ec1ec1e7dc4c2a765"}]},"branch":"refs/heads/stable/wallaby"},"6a04af2e76c8c8b8eab0ea084838aa2fcd599a11":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-12-30 17:31:18.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/37/868737/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/37/868737/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/37/868737/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/37/868737/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/37/868737/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/37/868737/3"}}},"commit":{"parents":[{"commit":"716b947c479e36ce2e9ec6d9e0bbab882f391892","subject":"Merge \"Add periodic-stable-jobs template\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/716b947c479e36ce2e9ec6d9e0bbab882f391892"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2022-12-22 14:46:04.000000000","tz":60},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-30 17:31:11.000000000","tz":0},"subject":"Make tox.ini tox 4.0.0 compatible","message":"Make tox.ini tox 4.0.0 compatible\n\n* removed skipsdist\u003dTrue to make sure placement available in the virtual\n  env. This was at least needed for the doc generation\n\n* doc target uses requirements (e.g. six, oslo libs) from requirements.txt\n  not just from doc/requirements.txt\n\n* passenv fixed as space separated list is not allowed any more\n\n* improved the doc generation command line to print all the errors\n  instead of stopping at the first one\n\nExtra updates in backport:\n* Updated constraints to stable\n* functional-py310 tox env was added in master commit but that is not needed\n  for stable branches so removing that in backports.\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/866943\nDepends-On: https://review.opendev.org/c/openstack/nova/+/868791\n\nChange-Id: I1de24b7a98f149fce33d4e5eca82460eea72140b\n(cherry picked from commit 8bb18ecb25ad1076aa594408c5ca360b9cf33c9e)\n(cherry picked from commit 9835a24f443e55211ea0f658fe801d9630149ba8)\n(cherry picked from commit 240e4b775ec8973c079416237b3f2ac1bccd5313)\n(cherry picked from commit 52ed8d8e013e83bf9c38c58d029e83264e5f8c95)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/6a04af2e76c8c8b8eab0ea084838aa2fcd599a11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/6a04af2e76c8c8b8eab0ea084838aa2fcd599a11"}]},"branch":"refs/heads/stable/wallaby","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
