)]}'
{"id":"openstack%2Fplacement~869525","triplet_id":"openstack%2Fplacement~master~I7e94d107962efdd7779733375327f5c8248c7087","project":"openstack/placement","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-01-13 06:24:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7e94d107962efdd7779733375327f5c8248c7087","subject":"Avoid rbac defaults conflict in functional tests","status":"MERGED","created":"2023-01-08 07:10:18.000000000","updated":"2023-01-13 06:25:43.000000000","submitted":"2023-01-13 06:24:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"869525-secure-rbac","meta_rev_id":"fbce5a1c593cf743a33dadc6701e863f6352768e","_number":869525,"virtual_id_number":869525,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-13 06:24:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2023-01-13 05:20:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2023-01-12 20:46:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2023-01-13 05:20:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2023-01-12 18:56:07.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-08 07:56:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-12 18:56:07.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2023-01-12 20:46:51.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"8248f7092d0fe13107e2c11a418826f789f18949","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-08 07:10:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab8031e11f3bd4cf8baf4a64831750182faec4e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-08 07:56:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6adfe62ac78049bd89d8737606ea20f4\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7a98808aadf74ca5accf9f8fb170c1da : SUCCESS in 45m 02s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b336924208184492be2584f16f507787 : SUCCESS in 43m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/405cf341e99c4f7488493ec5d23260d7 : FAILURE in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e78446f364164d108ce9985d06e22fa1 : FAILURE in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7da6dc852ea471e9363de1752ea9982 : FAILURE in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9df38f2652c94f00b16a7d903b3d7652 : FAILURE in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cd00f240d8e459b948d72fa8b3c47dc : FAILURE in 4m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/603d94b075e94b0598de4d552957d824 : FAILURE in 4m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8f2772d0d2c94435a43fde07f442c90c : FAILURE in 3m 48s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a68a3ada170349f6a9e052b96468ce34 : SUCCESS in 14m 13s","accounts_in_message":[],"_revision_number":1},{"id":"610a62fd75346a13215a89a6854749ade639b845","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-08 20:10:39.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"66979c86d7cb57ffa4c1e4f8e61e60e0913c6cc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-08 21:00:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b29791f9028481fb75f22e8a82b56b8\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b1bff08016814e73ae37d81022d577b8 : SUCCESS in 40m 12s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8010bd2d910c460080615b0711c0ed12 : SUCCESS in 45m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/087c2e51b62a48d3b306550727b3312c : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e42ecd158f584262add3ebaf284dd288 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e9fcff7a22944ad819d4ba322e4bcde : SUCCESS in 8m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f93c7a199f9a42c2a8518a90a2c9cf9a : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6b09735fbcb43faa30f461f1531e5a7 : SUCCESS in 9m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f9ee1ccd278347a29e6cc1546ddd2d77 : FAILURE in 4m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5f17446acbcc487eba7cf479f651811e : FAILURE in 3m 36s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a160fa9c90494180ac5de6db33aeb6b0 : SUCCESS in 20m 25s","accounts_in_message":[],"_revision_number":2},{"id":"16b2b0f8c19c5e8d593c326271ad8d85e1cef506","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-09 18:09:14.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"962b4de337aa2ec175ee26c54747b839a380bdd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 19:21:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/600044cd32a44b4381492c3349f2a228\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/bf08e2c8081a46c8afc77dcb5741cebc : SUCCESS in 1h 07m 32s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ed9e53bb0b14c308739a9ea3c2cc506 : SUCCESS in 56m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce90eccc8b814b849bed837d20f759a5 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/113f8f2c07074abdb3fe7b8996fd314f : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16f647412b094c54875d8c1f33ae8f8a : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62cb2c2387304992ba005495454991ee : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61f51c059bbb48acbefbb4af1559d4cf : SUCCESS in 8m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1edf3f5b849245eb8d97f84d30c03b00 : SUCCESS in 6m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/db4b15544e9f425b9dd09e8efcc532d9 : SUCCESS in 5m 36s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b2dba1c8c98049b7aede8498d4eadc73 : SUCCESS in 16m 47s","accounts_in_message":[],"_revision_number":3},{"id":"e524dc4dbfe020a911ecbca40fceb9cce315b490","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-09 20:41:21.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":3},{"id":"5bcb341b686cca590fc9902874808b568f3c9190","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-01-12 18:56:07.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"0e5fb6fd0a3d99fc220ec70bd3c825469b28c38f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-01-12 20:46:51.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"943bc03e9e4a0b5bbba33c60b69f352e85483154","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-01-13 05:20:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4513595b233f254c07277bfeb09d0066ef80febd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 05:21:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4f9cb11fcb5bc7c22271cf95512ef479ed2f4316","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 06:24:45.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e96645f832b440c19786a5b1daddef68\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e41b1494acd84ecabd6c6e96dda989a4 : SUCCESS in 58m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4464bf412f0b40d0b25a563ed64c0d1d : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce751588fb7e47dbbe74d0e00167ac2c : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1b4811e17424484b95f9688964301988 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13e26854906b479e9654ef1356dee116 : SUCCESS in 7m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ec4f64e6ba514f069845112b4091fb27 : SUCCESS in 5m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/29fbabcd9c5c497cbc7889033e753d0d : SUCCESS in 6m 16s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/829b8eb9f5c049f387cafb2014757877 : SUCCESS in 18m 50s","accounts_in_message":[],"_revision_number":3},{"id":"1851a6033fe2cf3d33a844ca595130b410f1172f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 06:24:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"fbce5a1c593cf743a33dadc6701e863f6352768e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 06:25:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/073f5e9f54cc4bfca3fa0b80526db4d4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/248c9e7e369f497daefbb8b2e03119f8 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ff8bee1fbc8151a399b096d8e4929ec923a9b02e","revisions":{"a54999b6b2c90ce2179e0584fafe5ce7658fc2f3":{"kind":"REWORK","_number":1,"created":"2023-01-08 07:10:18.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/25/869525/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/25/869525/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/25/869525/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/25/869525/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/25/869525/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/25/869525/1"}}},"commit":{"parents":[{"commit":"bacd2c039345dd12f22b0474b8a0f6a691a8ed98","subject":"Policy defaults improvement spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/bacd2c039345dd12f22b0474b8a0f6a691a8ed98"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-08 07:06:10.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-08 07:06:10.000000000","tz":-360},"subject":"Avoid rbac defaults conflict in functional tests","message":"Avoid rbac defaults conflict in functional tests\n\nRBAC defaults for enforce_scope and enforce_new_defaults for\nplacement service is False. Nova is setting them for Nova service\nto True in https://review.opendev.org/c/openstack/nova/+/866218\n\nPlacement tests fixture are used in Nova functional tests and\nsetting the rbac defaults causing placement to run with defaults\nof what Nova is setting. To avoid this conflict, in placement\ntests fixture, we need to set enforce_scope and enforce_new_defaults\nto the same value it is for placement service.\n\nNeeded-By: https://review.opendev.org/c/openstack/nova/+/866218\nChange-Id: I7e94d107962efdd7779733375327f5c8248c7087\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/a54999b6b2c90ce2179e0584fafe5ce7658fc2f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/a54999b6b2c90ce2179e0584fafe5ce7658fc2f3"}]},"branch":"refs/heads/master"},"e4437637848d6570e05183546a9f9a600c2fdcde":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-01-08 20:10:39.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/25/869525/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/25/869525/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/25/869525/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/25/869525/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/25/869525/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/25/869525/2"}}},"commit":{"parents":[{"commit":"e7dfe16cbbbfa49c20ca6213d8a98299bf50e3a2","subject":"Make tox.ini tox 4.0.0 compatible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e7dfe16cbbbfa49c20ca6213d8a98299bf50e3a2"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-08 07:06:10.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-08 20:10:39.000000000","tz":0},"subject":"Avoid rbac defaults conflict in functional tests","message":"Avoid rbac defaults conflict in functional tests\n\nRBAC defaults for enforce_scope and enforce_new_defaults for\nplacement service is False. Nova is setting them for Nova service\nto True in https://review.opendev.org/c/openstack/nova/+/866218\n\nPlacement tests fixture are used in Nova functional tests and\nsetting the rbac defaults causing placement to run with defaults\nof what Nova is setting. To avoid this conflict, in placement\ntests fixture, we need to set enforce_scope and enforce_new_defaults\nto the same value it is for placement service.\n\nNeeded-By: https://review.opendev.org/c/openstack/nova/+/866218\nChange-Id: I7e94d107962efdd7779733375327f5c8248c7087\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e4437637848d6570e05183546a9f9a600c2fdcde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e4437637848d6570e05183546a9f9a600c2fdcde"}]},"branch":"refs/heads/master","description":"Rebase"},"ff8bee1fbc8151a399b096d8e4929ec923a9b02e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-01-09 18:09:14.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/25/869525/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/25/869525/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/25/869525/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/25/869525/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/25/869525/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/25/869525/3"}}},"commit":{"parents":[{"commit":"1c8afcd3f10a2ace8fdbada4f758b27bb2774cec","subject":"Make tox.ini tox 4.0.0 compatible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/1c8afcd3f10a2ace8fdbada4f758b27bb2774cec"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-08 07:06:10.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-09 18:09:14.000000000","tz":0},"subject":"Avoid rbac defaults conflict in functional tests","message":"Avoid rbac defaults conflict in functional tests\n\nRBAC defaults for enforce_scope and enforce_new_defaults for\nplacement service is False. Nova is setting them for Nova service\nto True in https://review.opendev.org/c/openstack/nova/+/866218\n\nPlacement tests fixture are used in Nova functional tests and\nsetting the rbac defaults causing placement to run with defaults\nof what Nova is setting. To avoid this conflict, in placement\ntests fixture, we need to set enforce_scope and enforce_new_defaults\nto the same value it is for placement service.\n\nNeeded-By: https://review.opendev.org/c/openstack/nova/+/866218\nChange-Id: I7e94d107962efdd7779733375327f5c8248c7087\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ff8bee1fbc8151a399b096d8e4929ec923a9b02e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ff8bee1fbc8151a399b096d8e4929ec923a9b02e"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}]}],"submit_requirements":[]}
