)]}'
{"id":"openstack%2Fplacement~880623","triplet_id":"openstack%2Fplacement~master~I8724b0738f097739f55cd8566bb30b7ddbe154d9","project":"openstack/placement","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2023-06-14 21:05:10.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-06-14 22:48:12.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2023-06-14 22:48:12.000000000","reason":"Change was submitted"},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2023-06-14 22:48:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8724b0738f097739f55cd8566bb30b7ddbe154d9","subject":"db: Replace use of deprecated API","status":"MERGED","created":"2023-04-17 12:24:07.000000000","updated":"2023-06-14 22:49:03.000000000","submitted":"2023-06-14 22:48:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880623-sqlalchemy-20","meta_rev_id":"071bcbe140c06a9bbfbcf8f7594e6aea27b4af71","_number":880623,"virtual_id_number":880623,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-14 22:48:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":2,"date":"2023-06-13 22:45:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":2,"date":"2023-06-14 21:05:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":1,"date":"2023-06-14 21:05:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-17 13:40:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-17 15:00:55.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2023-04-17 15:00:55.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2023-04-17 15:00:55.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2023-06-13 22:45:08.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"}],"messages":[{"id":"19e62566b6180a207d420ee3d3feab76cbe02213","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-04-17 12:24:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"64f40a27fb0521478aff6824cb1cba5d1fd4b3ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 13:40:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed653c463d0a4a3eb38e9036da1b96c5\n\n- grenade https://zuul.opendev.org/t/openstack/build/d04428261a3c410591cf28c7b607816f : SUCCESS in 1h 00m 25s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/f00459d88f854fb59607c6d67c1dd82f : SUCCESS in 1h 14m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5db576cdb334444ab6056b8a04e910d8 : SUCCESS in 58m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12429d9ff9464b1e9e073203426fa04d : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ff3e5c3f97a45789fcaa69f0ec3ab07 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/865609869d4d4bdd90174f36e2c33e23 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b1cf0534176040f0b109c588071ee06f : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/530e931bfa7045b6ba3f925f61015f4f : SUCCESS in 6m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4064a5b508c84e23be927d699dc3b1f5 : SUCCESS in 5m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/805c601a4c4d44179eaeea7b23f51760 : SUCCESS in 6m 32s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e4ab44f377414e25b0abcb2c4441c1f5 : SUCCESS in 16m 32s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/b243ea1c5a454e528228ac8c64a14a62 : FAILURE in 4m 36s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/ecf91295d2fe4c13b17b1c3443a8a80d : FAILURE in 2m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/770e4ba59b4b49c8a41bbf6dfe7fc528 : SUCCESS in 51m 32s","accounts_in_message":[],"_revision_number":1},{"id":"ef2aa28ef43ef0f9d39069981fe4729839f89d64","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2023-06-13 22:45:08.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"416d8a972c3275c17ca291ef0b25c48e6517b9df","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-06-14 21:05:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cb78dd82dd24d4ffa957e8a02f468f630f3104ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 21:05:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fc09c5a3cf476218f9d5f43861931cef3131b2e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 22:48:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/850cd2abfe5b463aa12481931cb5213f\n\n- grenade https://zuul.opendev.org/t/openstack/build/deca1e00e75b4c84b2fcd2076fd59623 : SUCCESS in 1h 04m 51s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/764e86c2629c4d49a4733bf72a691a98 : SUCCESS in 1h 35m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98358eebeba243ea9f49bb3918567cee : SUCCESS in 1h 03m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58cef10eab574b708bab4e9b3cf747c7 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11e98512551d433f861c7897f4f7e60d : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/142d9ada1af6429990fdd1128c1bd6e5 : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac3ca7de8bf04efeb715080389169a1b : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80e7c84f92414bb2b3974fa08125ef7f : SUCCESS in 7m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/76c8d9a2ab1b4f07810c5a0777be1f47 : SUCCESS in 5m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/494f5d75c86345f48aef15712a38c942 : SUCCESS in 5m 27s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d08ac8fe6cc44dfc90f592300c43bff8 : SUCCESS in 17m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c529bd338ea4a40b10ea2ca7c66b4e7 : SUCCESS in 41m 15s","accounts_in_message":[],"_revision_number":1},{"id":"e3ac5a180a85aeceb14a6daeffb9f673d662be31","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 22:48:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"071bcbe140c06a9bbfbcf8f7594e6aea27b4af71","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 22:49:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/713247cf9bf64f14984744348c316147\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c16b28a73594787b113cfb048e43c8c : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f71603d69575a5a271c285c61f26bd89bed1247b","revisions":{"f71603d69575a5a271c285c61f26bd89bed1247b":{"kind":"REWORK","_number":1,"created":"2023-04-17 12:24:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/23/880623/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/23/880623/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/23/880623/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/23/880623/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/23/880623/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/23/880623/1"}}},"commit":{"parents":[{"commit":"d7e9a330f03c436544285db1742fdff2727e79c9","subject":"Merge \"Db: Drop redundant indexes for columns already having unique constraint\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d7e9a330f03c436544285db1742fdff2727e79c9"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-04-17 10:04:10.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-04-17 10:47:55.000000000","tz":60},"subject":"db: Replace use of deprecated API","message":"db: Replace use of deprecated API\n\nWe were missing a number of users of SQLALchemy APIs that were either\nchanged or removed in 2.0. These were yielding the following errors in\n2.0.\n\n  TypeError: MetaData.__init__() got an unexpected keyword argument\n  \u0027bind\u0027\n\n  TypeError: TableClause.insert() got an unexpected keyword argument\n  \u0027values\u0027\n\n  TypeError: TableClause.update() got an unexpected keyword argument\n  \u0027values\u0027\n\n  AttributeError: \u0027Engine\u0027 object has no attribute \u0027execute\u0027\n\nResolve these issues. We will fix the test class issue in a follow-up.\n\nChange-Id: I8724b0738f097739f55cd8566bb30b7ddbe154d9\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f71603d69575a5a271c285c61f26bd89bed1247b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f71603d69575a5a271c285c61f26bd89bed1247b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
