)]}'
{"id":"openstack%2Fplacement~888281","triplet_id":"openstack%2Fplacement~master~I7fb1538174c847dd4b9f1d599f1895982f0d1ec2","project":"openstack/placement","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2023-11-06 12:38:28.000000000","reason":"Change was abandoned"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2023-08-17 13:55:21.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-09-27 15:01:46.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2023-11-06 12:38:28.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7fb1538174c847dd4b9f1d599f1895982f0d1ec2","subject":"db: Workaround SQLAlchemy 2.x bug","status":"ABANDONED","created":"2023-07-12 14:09:14.000000000","updated":"2023-11-06 12:38:28.000000000","total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8cdee6da2c6e369f21acd9bd0bf0580961dcee5d","_number":888281,"virtual_id_number":888281,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-09-27 14:14:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":2,"date":"2023-09-27 12:37:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2023-09-27 12:37:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-12 15:36:25.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2023-07-12 15:36:25.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2023-07-12 15:36:25.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2023-07-12 15:49:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-27 12:35:33.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"}],"messages":[{"id":"6a067b3f3264116494b5ff1a8eb59dc9bfabfcef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-12 14:09:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51097710b94ceb395a4fc87c51421da552c54e92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 15:49:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/89a6c92efb7e4dc9b23c062af551bdac\n\n- grenade https://zuul.opendev.org/t/openstack/build/7bfce34946df430da05c5dbb7002c28a : FAILURE in 39m 52s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/33e67cf315b84cc295d55e0b9c4badb0 : SUCCESS in 1h 19m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/69c00122a0e44f4c96bfe116d9768333 : SUCCESS in 59m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11a4031d3b7048439cdcfcb8477a203f : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dd6067e1bdd4bdfb478b76ca9c03ac5 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1663a60d38744ccda92ecf34c2a54938 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d91b95b3f5174ba790eb9c9e122d2dbc : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a5ad417ac8b44d59d5ac97cdf014f36 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a71fca0c7d649888fe3e28cfec83395 : SUCCESS in 7m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b4d4e39040534905b4ad43484b4bb412 : SUCCESS in 5m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3c437985cb30495781f497920b4df597 : SUCCESS in 5m 17s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/19c59fec260043b0bdc493a3c99b57f2 : SUCCESS in 19m 49s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/6a4c38e652ed4c7b86ed260ab7304063 : FAILURE in 4m 01s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/7bf32f54098e4062af8dd27e0908c5f5 : FAILURE in 2m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09fa238695484879a4e79f12a4cec5b8 : SUCCESS in 1h 05m 50s","accounts_in_message":[],"_revision_number":1},{"id":"19993f1e629a3b5a160d99b28d710970476d6eca","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-17 10:55:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2f0c6e6d5cdd22b8e9c4d5073d74ce9f86798361","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 12:38:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f555c6617f5b4d549abe4b18335305ea\n\n- grenade https://zuul.opendev.org/t/openstack/build/7af6e12af77049ac85f871abe383186a : FAILURE in 51m 44s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/1009ebada9a04e308d139ac929e65452 : SUCCESS in 1h 36m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/91323c022bab4193a18cd33bc97780af : SUCCESS in 57m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42c59f1005e44cf5a1f2062e476ab66c : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dd8249255d34d47a460156a8994abbf : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27e28cce3aea4df3bee7f93256354a5a : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d747ef40ff0a4efaa9c7393cdadf8112 : SUCCESS in 5m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8937520d6cbb46c387b723ede3afb38d : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/422f9fbffbe34585961d5b2722b4485a : SUCCESS in 6m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/17ad9c9f518749f7aa2371530ddff2f3 : SUCCESS in 5m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2b6c07a1ee364eafb78b924e2708b7d1 : SUCCESS in 5m 22s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/22e771fafc13422687310dc115309e5c : SUCCESS in 21m 11s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/3819b561c63d4cbbb86a2b403caff65a : FAILURE in 3m 41s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/d2a58a01e1eb4155b92223c10a84ff4c : FAILURE in 2m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8496b98690aa4fdaa94844ae074af2d6 : SUCCESS in 1h 05m 08s","accounts_in_message":[],"_revision_number":1},{"id":"df1744a04ddd9655a050933195b641e339d8f3b7","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-08-17 13:55:21.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8298c4ecd48f16605be72e3c689d1119e060b68d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-09-13 09:58:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"80c7832923e523a10421298ecb20a3e52a44dcb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 11:36:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3caaf9e064d4b90aa016b8d085b66f2\n\n- grenade https://zuul.opendev.org/t/openstack/build/606cd6d6aa204cfd85278b35e9edf8b6 : FAILURE in 46m 19s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/3530723c97574b1b840920af1890fb58 : SUCCESS in 1h 30m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cbbf0459de8a4d14b84d2fe80257b68b : SUCCESS in 41m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20d06f2125554c42be57d49edcbef474 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a4fd1e4a11a40d1a54fb01dc91366dd : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5d3f2eaebe045f6801f18f281423841 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42e4b2178b3f4571941769db665479ba : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78fc7f31a48d4deda071f586e6512f81 : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eccbddebd61e47edbdf2cceb9d9af993 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e294979ed1043ae939bbc1d75e298c4 : SUCCESS in 7m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b6cdfbec2be944f6ad0b2d8965a41c11 : SUCCESS in 4m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/62346328fdd04061a519c07cea5a6b8d : SUCCESS in 5m 10s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b8efa7ac07f14530a0ecbc219521d35d : SUCCESS in 20m 17s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/445490ca7d2f4dde8600b791c7ea4543 : FAILURE in 3m 37s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/08b51f0d258244a09c079fbb0ef005fa : FAILURE in 2m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/281ea6005cc147bb9f497a4d5c1c3291 : SUCCESS in 1h 01m 52s","accounts_in_message":[],"_revision_number":1},{"id":"7ad498a6fc8af37e2d782a1219d5a529161de4c5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-13 14:21:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9679737a533b9ed81a3c0a35bae707aaca74bb00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 15:39:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a94c16a5114440799ff5b77f36d3812e\n\n- grenade https://zuul.opendev.org/t/openstack/build/aa2b9cfe1de9498abffdbeec0ac2cb49 : FAILURE in 44m 25s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/9f692b50b14f46adae74c33693950c01 : SUCCESS in 1h 16m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2acb058f6f2c4b9cbab856fd253abc27 : SUCCESS in 58m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df3038fe45e148629cadff6abaa18a97 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc4c61dbff084be8818441545e008da3 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0fdf0287db1a43c5bcfb50c9c662a8d7 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41af0427af984a83af94c250f4fb7f6c : SUCCESS in 6m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78d3fc725e324839ab00ef2c2fd5e73b : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/34c6f4acb6974e86ad3d4da425d68c87 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0948a856a31048d09619722a739496f3 : SUCCESS in 8m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f7edec9e48ab4c05908a509bb212c1b2 : SUCCESS in 5m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/28dc5a9bbee04603abcf93893827d8e8 : SUCCESS in 5m 45s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9312f778aecd4d4db65bb22a496b282c : SUCCESS in 19m 22s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/727a858044fd4e91a6afd4aa11106c6c : FAILURE in 4m 12s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/667df10eaed14ae98fc67179ae623d09 : FAILURE in 3m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/344f5a22938848fda0314019bea00b8c : SUCCESS in 1h 01m 23s","accounts_in_message":[],"_revision_number":1},{"id":"e3604558bd3b0c8c14fde1328adbe0215e3fca71","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-27 12:31:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c3c215f023c32ab696af6d8483ecd2afc1e9ff25","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-09-27 12:35:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1b141c7480e59e7e5e6dfa317f391bfed2d53d20","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-09-27 12:35:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fb1125c5d7ed0f0a863f849166e3a7d4ed5ca7d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-27 12:37:09.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c09520fee064bc3d1605e188edc4a0aa69578647","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 14:14:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/89c446892dbe416eb45e62f83a967809\n\n- grenade https://zuul.opendev.org/t/openstack/build/b6d0d8d678614ff9a6626086a41dff67 : FAILURE in 50m 38s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/c7879adbb3a14bde957045ba0e4d3e01 : SUCCESS in 1h 36m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b97c9e10b07e43ce8a5305fcf12ad18f : SUCCESS in 1h 08m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c4b99ddcb424126b9357902cec83a2f : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78209f4c278d41cc98b91f968088d39d : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4072bb604ed4c10be851e10784d0d68 : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e9ca192d44043858bdc2d72641edc82 : SUCCESS in 5m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba572f007e1f432382c181d90ffc34bb : SUCCESS in 5m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/877bd67866804c56884a1f38324086fe : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b2ef87e7ee3496aaac18167bd42e3d8 : SUCCESS in 8m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/45030d2989e5428a99ea54db9bccca55 : SUCCESS in 6m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/74e40014ea6f4c438c57686ce8f832f3 : SUCCESS in 6m 20s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bb11adaf02034977b6c6ad067a7a5682 : SUCCESS in 18m 47s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/24a30694cddf4bfd80dc761800204220 : FAILURE in 4m 31s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/eff8af6228af4f48aa6684e02a02989c : FAILURE in 3m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab7eb76c5ca0464e96e06eb8f80ca819 : SUCCESS in 44m 47s","accounts_in_message":[],"_revision_number":2},{"id":"01e78e352e74e638913de54571ec7637c0c27fd6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-27 15:01:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8cdee6da2c6e369f21acd9bd0bf0580961dcee5d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-06 12:38:28.000000000","message":"Abandoned\n\nThis was fixed by SQLAlchemy 2.0.19 and my query cleanup here isn\u0027t supported by MySQL. I could revert the cleanup parts of this but it hardly seems necessary now. Abandoning instead.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3e53aaa2c6b83fd875bf94cb3cfa7c7385bfd647","revisions":{"14456df477ed5d73aca5585f71e87f5af2d8f9b2":{"kind":"REWORK","_number":1,"created":"2023-07-12 14:09:14.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/888281/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/81/888281/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/81/888281/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/81/888281/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/81/888281/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/81/888281/1"}}},"commit":{"parents":[{"commit":"07b2a3a249f5fe2b82909bb1dc2f514132edbff9","subject":"Merge \"tests: Warn on *any* SAWarning warning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/07b2a3a249f5fe2b82909bb1dc2f514132edbff9"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-07-12 13:59:05.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-07-12 14:08:56.000000000","tz":60},"subject":"db: Workaround SQLAlchemy 2.x bug","message":"db: Workaround SQLAlchemy 2.x bug\n\nThe Query API is legacy in SQLAlchemy 2.x and is effectively a shim over\nthe newer \u0027select()\u0027 API. Unfortunately, there is currently a bug [1] in\nthis shim which we\u0027re seeing in early gating jobs for SQLAlchemy [2].\nWhile that bug will eventually be addressed, it will take some time and\nstill leaves using a legacy API that we should probably migrate off of.\nDo just that, allowing us to avoid the wait for a fix and getting us\ninto a more SQLAlchemy 2.x friendly place. We use the opportunity to\nsimplify the query somewhat, removing a redundant subquery, and to\ncorrect a clearly errant docstring.\n\n[1] https://github.com/sqlalchemy/sqlalchemy/issues/10098\n[2] https://review.opendev.org/c/openstack/requirements/+/879743\n\nChange-Id: I7fb1538174c847dd4b9f1d599f1895982f0d1ec2\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/14456df477ed5d73aca5585f71e87f5af2d8f9b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/14456df477ed5d73aca5585f71e87f5af2d8f9b2"}]},"branch":"refs/heads/master"},"3e53aaa2c6b83fd875bf94cb3cfa7c7385bfd647":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-09-27 12:37:09.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/888281/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/81/888281/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/81/888281/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/81/888281/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/81/888281/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/81/888281/2"}}},"commit":{"parents":[{"commit":"141780c0ce0836cbb258a9520dfc26648bb133c5","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/141780c0ce0836cbb258a9520dfc26648bb133c5"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-07-12 13:59:05.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-27 12:37:01.000000000","tz":60},"subject":"db: Workaround SQLAlchemy 2.x bug","message":"db: Workaround SQLAlchemy 2.x bug\n\nThe Query API is legacy in SQLAlchemy 2.x and is effectively a shim over\nthe newer \u0027select()\u0027 API. Unfortunately, there is currently a bug [1] in\nthis shim which we\u0027re seeing in early gating jobs for SQLAlchemy [2].\nWhile that bug will eventually be addressed, it will take some time and\nstill leaves using a legacy API that we should probably migrate off of.\nDo just that, allowing us to avoid the wait for a fix and getting us\ninto a more SQLAlchemy 2.x friendly place. We use the opportunity to\nsimplify the query somewhat, removing a redundant subquery, and to\ncorrect a clearly errant docstring.\n\n[1] https://github.com/sqlalchemy/sqlalchemy/issues/10098\n[2] https://review.opendev.org/c/openstack/requirements/+/879743\n\nChange-Id: I7fb1538174c847dd4b9f1d599f1895982f0d1ec2\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/3e53aaa2c6b83fd875bf94cb3cfa7c7385bfd647"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/3e53aaa2c6b83fd875bf94cb3cfa7c7385bfd647"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
