)]}'
{"id":"openstack%2Fplacement~896650","triplet_id":"openstack%2Fplacement~master~I31123c70c9664844181eeaaf9d0846a319af7c2c","project":"openstack/placement","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-10-26 17:31:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I31123c70c9664844181eeaaf9d0846a319af7c2c","subject":"db: Wrap raw SQL query in sqlalchemy.text","status":"MERGED","created":"2023-09-27 12:37:09.000000000","updated":"2023-10-26 17:32:22.000000000","submitted":"2023-10-26 17:31:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896650-sqlalchemy-20","meta_rev_id":"195faade56921fe8f33ef7a0fbabf19f1fe5c777","_number":896650,"virtual_id_number":896650,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-26 17:31:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"recommended":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"all":[{"value":1,"date":"2023-10-01 14:46:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-26 15:54:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2023-09-29 15:48:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-26 15:54:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-27 14:24:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-29 15:48:07.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2023-10-01 14:46:03.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2023-10-26 15:54:12.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"ff53a644f1dd11742d8d567bb8cf28efb9a032bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-27 12:37:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4fcfc86d0be5d740ca61c2cb290540be58f6371","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 14:24:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7310e0326d24291afc9e33495b68f8d\n\n- grenade https://zuul.opendev.org/t/openstack/build/366239e1701e4a928dffefae798893c0 : FAILURE in 56m 53s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/24c64ccdcdbb44bb8798e8731ae545bc : SUCCESS in 1h 41m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/24647d25d0a44535a08631989f6b056e : SUCCESS in 1h 16m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6dc3dcbf460b40798f012106622b1aaf : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b0f8be74c814da38b163eb8ea5a4cd6 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc40583bc3cc44b7b141a6c66940acf2 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb8aa5b58823469084c56f4cc47e44d6 : SUCCESS in 6m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c9267931562349bdbdfe871d3e3976ef : SUCCESS in 6m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cc529205c394441ca1488f08b54880ec : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fb1c515ce5f4e63a2adf8068533ac88 : SUCCESS in 7m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/aec5dbef70114f3b834f7160bbd73215 : SUCCESS in 6m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/616d4aac240f4502b77a5e95050307e2 : SUCCESS in 5m 53s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7e6fa35b9315424dbf0c8f07aef10299 : SUCCESS in 22m 24s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/414766626c314977a1d58e75f9a558f7 : SUCCESS in 20m 33s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/f5a1ecf806934ffc89b68feebe6f0ff7 : SUCCESS in 9m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d29cb5f7a09b458eaf4a6ca0b70cb822 : SUCCESS in 1h 07m 22s","accounts_in_message":[],"_revision_number":1},{"id":"edf58b45b8765f0a36cb24bc9b0d90d4ed17203f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-27 15:01:53.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6835e20b57e9b12cce449e02538733d7c525535f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 16:16:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/716e611a4f444dc5a1d59c730560d012\n\n- grenade https://zuul.opendev.org/t/openstack/build/64d13ffb75fd4419b0428b7b5ce35702 : SUCCESS in 57m 46s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/f33b2937f66c4e1bb4c5dd76a7970783 : FAILURE in 1h 04m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/661815bed9944cdfbf44f070362bbd1b : SUCCESS in 1h 08m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/848ecb630beb48a9918960db0fc924c7 : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46833be3a18f42e686971aea7a2b4b39 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e92565b9ffe4b2dab2d9419f9f95811 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62cc3af437014be4896f7458b4175a96 : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/936dade7c80c4f008567d695c486fba0 : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9610a7db67f947c3b36d873a17fb1cf2 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/209200090bdd446793370872ab113232 : SUCCESS in 6m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/453f0a07c3634628a1d49113ea42c75b : SUCCESS in 5m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/57fa6f993573421f8a3ea7b27d1d2829 : SUCCESS in 5m 43s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/32ab4511ab014a9f99314ebf966c9191 : SUCCESS in 19m 53s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/81a0f2d80d8e4dcf9789f6686e27e46e : SUCCESS in 31m 33s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/3b25553e24ec42dda40ff36cbe5c9330 : SUCCESS in 13m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/644da521d65d4fe8bcfafe0b2e1d8320 : SUCCESS in 55m 38s","accounts_in_message":[],"_revision_number":2},{"id":"d2e602c2f254b0b1b03ff902be63b01a8bcd31cd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-28 11:23:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"54b1c7447f86e04c398518829730839ca23c3830","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 12:46:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3e8bbda88334c7c98e020e39f1f8d4a\n\n- grenade https://zuul.opendev.org/t/openstack/build/f4458eb8f448400a8d214367ce697d46 : SUCCESS in 49m 01s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/2cf471f7e10a4723a73e7e4192340044 : SUCCESS in 1h 20m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37ae803b27a84f408146e1360ab76b74 : SUCCESS in 43m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fec005f0e0a4804aca9b8034dd6ab83 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f37a865b63204c7c8fac31d0d638c9f0 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2b870cf456c49ea90ddd91546827adf : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32179c906a934743bbbd7e60cddaea92 : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/595a6833c94f4daeb7e5c9af54c818bb : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8e9d3c20260749a4bc7fdfa899485017 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d116877fe0644efab1f038695ebf11e : SUCCESS in 7m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a5a04d8edf5d4a9d836894eb48f2fd65 : SUCCESS in 5m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2acfc7a08a8a4be8863ece3095ed9e23 : SUCCESS in 5m 41s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9a72c5ae536a4a71bf08e4cf7236064f : SUCCESS in 24m 12s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/949b937efc0f4fa080b678d840f2af4b : SUCCESS in 23m 13s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/e643d8c82a284df0984cd88ce8522a4e : SUCCESS in 10m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c9e0bcef55f4349b4c5f9362d4cadb0 : SUCCESS in 1h 06m 37s","accounts_in_message":[],"_revision_number":2},{"id":"ebcc0951eb283521088dc04f5a31aac84317b7e3","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-09-29 15:48:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c8928399f27fdaed831da97506f8afe9977e8148","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2023-10-01 14:46:03.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"46e83c46f741441ee84d930f61b6ccae78737091","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2023-10-26 15:54:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"798b57176a998eafa78b69964c2098706dec0dae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 15:54:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9de398e1ea15b7568ac349fb9b6f01721c81421b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 17:31:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f04204bd81a405eb7990af9edc746be\n\n- grenade https://zuul.opendev.org/t/openstack/build/e4ca9542f45b4bd2bc6490d8b27cc1ea : SUCCESS in 52m 06s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/81ec6f893fe44a7d96c4139cf804251c : SUCCESS in 58m 10s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a1089bc3ff2a488d93f165f6f605e569 : SUCCESS in 1h 15m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c7cc493603c4fba94e0a9775cdf875e : SUCCESS in 1h 09m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50d9c6d587474794bf942d0cb87cd732 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2fac69655a8249c194f8bcafa29b36dc : SUCCESS in 6m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f28ae80df91c4a27aea02604e762fc97 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81dee0caa1b9496391c6b54074dfdbd2 : SUCCESS in 8m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7fa3f150c9c84266aa821145cbfd36c5 : SUCCESS in 7m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5804a8c93ce94f89a972cb667debc307 : SUCCESS in 7m 25s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dabf017b0052464b98c69e7c83b68aa7 : SUCCESS in 19m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21e97f4b786b4454b1b8ca884720e56d : SUCCESS in 1h 02m 53s","accounts_in_message":[],"_revision_number":2},{"id":"275973559f547a9165e72171b05ee82956111009","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 17:31:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"195faade56921fe8f33ef7a0fbabf19f1fe5c777","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 17:32:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4d4cc02389b4e68917d21ceab25119f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97f8213b91e748e19ea1beb3afca9fae : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d638b011bb5fc24a22042041baafba590685a098","revisions":{"96df54c0eab66473b9d1f38c435f26c9434c630c":{"kind":"REWORK","_number":1,"created":"2023-09-27 12:37:09.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/50/896650/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/50/896650/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/50/896650/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/50/896650/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/50/896650/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/50/896650/1"}}},"commit":{"parents":[{"commit":"3e53aaa2c6b83fd875bf94cb3cfa7c7385bfd647","subject":"db: Workaround SQLAlchemy 2.x bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/3e53aaa2c6b83fd875bf94cb3cfa7c7385bfd647"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-27 12:35:10.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-27 12:37:01.000000000","tz":60},"subject":"db: Wrap raw SQL query in sqlalchemy.text","message":"db: Wrap raw SQL query in sqlalchemy.text\n\nAnother change in SQLAlchemy 2.x.\n\nChange-Id: I31123c70c9664844181eeaaf9d0846a319af7c2c\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/96df54c0eab66473b9d1f38c435f26c9434c630c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/96df54c0eab66473b9d1f38c435f26c9434c630c"}]},"branch":"refs/heads/master"},"d638b011bb5fc24a22042041baafba590685a098":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-09-27 15:01:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/50/896650/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/50/896650/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/50/896650/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/50/896650/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/50/896650/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/50/896650/2"}}},"commit":{"parents":[{"commit":"141780c0ce0836cbb258a9520dfc26648bb133c5","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/141780c0ce0836cbb258a9520dfc26648bb133c5"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-27 12:35:10.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-27 15:01:53.000000000","tz":0},"subject":"db: Wrap raw SQL query in sqlalchemy.text","message":"db: Wrap raw SQL query in sqlalchemy.text\n\nAnother change in SQLAlchemy 2.x.\n\nChange-Id: I31123c70c9664844181eeaaf9d0846a319af7c2c\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d638b011bb5fc24a22042041baafba590685a098"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d638b011bb5fc24a22042041baafba590685a098"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
