)]}'
{"id":"openstack%2Fplacement~904618","triplet_id":"openstack%2Fplacement~master~Id0f37bebfb5298c5a84aa8f46f9b2d161b1f81ca","project":"openstack/placement","branch":"master","topic":"py3.11","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-03-15 10:01:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id0f37bebfb5298c5a84aa8f46f9b2d161b1f81ca","subject":"Update python classifier in setup.cfg","status":"MERGED","created":"2024-01-04 03:27:39.000000000","updated":"2024-03-15 10:02:39.000000000","submitted":"2024-03-15 10:01:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904618-py3.11","meta_rev_id":"5ab15891d9fefc13d0dc6ce5b317b12bc4b31f58","_number":904618,"virtual_id_number":904618,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-15 10:01:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"recommended":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-01-12 16:49:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2024-03-01 13:15:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":1,"date":"2024-01-09 05:12:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-03-15 08:17:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-03-15 08:17:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-04 05:03:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-09 05:12:15.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-01-12 16:49:03.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2024-03-01 13:15:02.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2024-03-15 08:17:32.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"}],"messages":[{"id":"6822d3edacfa6e7433bed6fda7cf6a813e5fe1b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-01-04 03:27:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61f3bd0d8af9e15cc665865a8b92560679b3e6fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 05:03:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71fe04ba1da0402aab6fdb7b9c1a0cb2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/339a24ac1fa6443cae71f84ee9e9dfd5 : SUCCESS in 2m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/aff4966bca0e4c778864538af89216f7 : SUCCESS in 1h 00m 57s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6894b3d352634cd680fa733ac718e097 : SUCCESS in 57m 22s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/5d58dc0eb7c244789b86e29d11155d61 : SUCCESS in 1h 29m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3609c30249554683be248c3b5aef868f : SUCCESS in 1h 09m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38d3ee389a6b44a8b41addf76cfe0b83 : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/511608c465c940999129bbca25df5a1f : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10bd17be477843c5a2f8a8adc1681c4a : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3d096630eea841a3af7de9f7e4a5702d : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34c28dcd91bd4beba3ddfbf9e82a1942 : SUCCESS in 8m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e96225359df54e14a61833056a5c83d0 : SUCCESS in 6m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8a5f179634ae41d897ed79ba87462da8 : SUCCESS in 5m 42s\n- placement-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ca97f51d4dc941259d6df40704d511c0 : SUCCESS in 9m 43s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2a0493d11106430da158fc24115c8c7f : SUCCESS in 18m 29s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/b9676bcf106a4f21a406efadb8b6f36e : SUCCESS in 18m 47s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/7052ac652ce847dc9a218239992e71f1 : SUCCESS in 8m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b4f64031ad246c6b84f4d72e719b676 : SUCCESS in 47m 48s","accounts_in_message":[],"_revision_number":1},{"id":"d52dc3a7cd84626d1033b7d3b36a7157b3e7f439","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-09 05:12:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f805fee8c443a0df5dff27ee2a98672739ff5851","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2024-01-12 16:49:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"55a57e13ab7db1d18f05f1eb0176f8e3c3145a5a","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2024-03-01 13:15:02.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0c7611df074472187c6cfc9b92b1a4c1ded08e2a","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2024-03-15 08:17:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8dce7167724d4883fb8ce54be254285516bdc35f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 08:18:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fe550c316bc5ee6af3af066499ff8b2d4bb36e00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 10:01:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71bcde82af36460aae12c4b61709fb39\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6a3eb8c185ec42338f9a41648d8d7d26 : SUCCESS in 1m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/58329a0bdafa4eff844f3ff6a655f2a2 : SUCCESS in 46m 14s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/1ed14bd3e0624a398a414c424db72604 : SUCCESS in 1h 33m 06s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/88089f72886544c58e6326d16baa30fc : SUCCESS in 1h 00m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00fdef672bf84f0395563e6fec001336 : SUCCESS in 52m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ded34879e704135810709121bca1d91 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43347a5accf0494f9585f478e206f009 : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/515cbb76470f45d39ef4246902ab821a : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/629838a0a1684b7d9de109555a2b4862 : SUCCESS in 7m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0c85215a2b984053bc3ce65019ec9f43 : SUCCESS in 7m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/148af0cb2c7d4b9f810e21d7b94773a5 : SUCCESS in 5m 44s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b6ab04e7c1c54d248720c8721897ead1 : SUCCESS in 19m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/381a9beb303a4ca88a66594856e022ff : SUCCESS in 1h 02m 57s","accounts_in_message":[],"_revision_number":1},{"id":"bf240f289346a1ba41bc9717c60ea80c5f03ae6d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 10:01:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5ab15891d9fefc13d0dc6ce5b317b12bc4b31f58","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 10:02:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a03c89c8c9614a4c8076a8110a4b2579\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f770be68ee54b8c8515a7c6a0d591db : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1001769dfbf58cd93c926a5da720b62bc9c1730f","revisions":{"1001769dfbf58cd93c926a5da720b62bc9c1730f":{"kind":"REWORK","_number":1,"created":"2024-01-04 03:27:39.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/18/904618/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/18/904618/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/18/904618/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/18/904618/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/18/904618/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/18/904618/1"}}},"commit":{"parents":[{"commit":"e51085604ca57d78469c4497289a687eade6a5e6","subject":"Merge \"Add job to test with SQLAlchemy master (2.x)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e51085604ca57d78469c4497289a687eade6a5e6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-01-04 03:22:50.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-01-04 03:22:50.000000000","tz":-480},"subject":"Update python classifier in setup.cfg","message":"Update python classifier in setup.cfg\n\nAs per the current release tested runtime, we test\ntill python 3.11 so updating the same in python\nclassifier in setup.cfg\n\nChange-Id: Id0f37bebfb5298c5a84aa8f46f9b2d161b1f81ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/1001769dfbf58cd93c926a5da720b62bc9c1730f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/1001769dfbf58cd93c926a5da720b62bc9c1730f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
