)]}'
{"id":"openstack%2Fplacement~907742","triplet_id":"openstack%2Fplacement~master~Idbe56fd0eaa2e4b04ce64051bd706ad7757cddf2","project":"openstack/placement","branch":"master","topic":"slurp-job","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-03-15 09:58:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idbe56fd0eaa2e4b04ce64051bd706ad7757cddf2","subject":"Add upgrade job from 2023.1","status":"MERGED","created":"2024-02-05 14:03:27.000000000","updated":"2024-03-15 10:01:06.000000000","submitted":"2024-03-15 09:58:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907742-slurp-job","meta_rev_id":"2eaf1520bcd67773fb62848b88634ada49890082","_number":907742,"virtual_id_number":907742,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-15 09:58:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"recommended":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"all":[{"value":1,"date":"2024-03-01 13:14:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":2,"date":"2024-02-07 12:21:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-03-15 08:15:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-03-15 08:15:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-05 15:16:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-07 12:21:49.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2024-03-01 13:14:30.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2024-03-15 08:15:05.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"}],"messages":[{"id":"d28e37354cbfe60bc61555460db131b57e69fbf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-05 14:03:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d37e518d8d019d0b1679a7ef5f4b60670bf7756c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-05 14:05:44.000000000","message":"Topic set to slurp-job","accounts_in_message":[],"_revision_number":1},{"id":"a4bec842a1f7200b3d4cc0d65e709f9b7d5c4b37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 15:16:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/331ed1964cd64f4f855d11f5c81d5292\n\n- grenade https://zuul.opendev.org/t/openstack/build/11008c4233d246cfb27916cf1bbb5516 : SUCCESS in 52m 06s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f3afd07f40ef493a985e79a4f80cd2ad : SUCCESS in 53m 23s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/4cb0cca045b44c13b0f25e823961636b : SUCCESS in 1h 09m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aff518d97eb1466bb204c000572589ff : SUCCESS in 23m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bd63557567f4e499b5bed33be90fddb : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3498256e5b74886a10b16df7140bcef : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc528ac5dfa4461b9934a1a1afcd7c19 : SUCCESS in 4m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d3b044f938f24960ba28e39ee5327eac : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2689fd2ad7542bcbecb2e2e82183d70 : SUCCESS in 7m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c9705f711ebe48e2aa0872073c54f88a : SUCCESS in 6m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c98f1a072edb4f7da42a845f410a4d5b : SUCCESS in 5m 57s\n- placement-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/eada0ef8cdc046eaad4245160f3f2ecf : SUCCESS in 7m 25s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6749c13a6e354f49a7985b8484ed2825 : SUCCESS in 19m 40s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/bf33ca6d8b344de987a11a7b511b66d4 : SUCCESS in 20m 44s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/af6a572d18094a4982b131060f50f11e : SUCCESS in 8m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67f6c59ff68f4c008c1081b078af4a62 : SUCCESS in 49m 57s","accounts_in_message":[],"_revision_number":1},{"id":"8c253fa5ffb8a4c1fb79fa1c9e223a47fe81d43f","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-02-07 12:21:49.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"638a800a90c4d45407e57aec8f966c45438f1d1b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-08 04:42:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"39311e73d9bf65e7b11bb703c3b0710c255d1073","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2024-03-01 13:14:30.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5ed34a771a83e198e270210744d8edac59a7c0b2","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2024-03-15 08:15:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bbbc4be8320cbcc6a79d8aeb5a1b302749c00671","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 08:15:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c829b7e7fb61f8fe8e1af1f4747732e559d32262","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 09:58:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a09744baf9d4554bc8ff6fe96bc1874\n\n- grenade https://zuul.opendev.org/t/openstack/build/97ee55734eb84a0eb9e26e1abd21aa12 : SUCCESS in 1h 01m 39s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/eed14e4db7d444d5b8f07323946c06fe : SUCCESS in 1h 37m 23s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ca495984b0ad456ea5faade4737e2442 : SUCCESS in 54m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e696d5585da9488a8f2a8fc7d534b60e : SUCCESS in 46m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7907cc8b4dc44870a1ed07a5a3250bfd : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/980f38905c3d4510b1680968edd1549d : SUCCESS in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6eeb8cfd7d93407b9724791a4134fc95 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16a79417792841989e2c1e10457d7d26 : SUCCESS in 7m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e5bcae842be64d3aab4d04ac308ab811 : SUCCESS in 6m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a31c02aa7e1143f89e26cd7af20b2082 : SUCCESS in 5m 02s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ab1bd67a9aed4113bf32c707a704644c : SUCCESS in 17m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0cb579502c304a2fab25f8dae6126f63 : SUCCESS in 49m 13s","accounts_in_message":[],"_revision_number":1},{"id":"8c1e084259b00978defe69c61dddd87bb0868a2d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 09:58:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2eaf1520bcd67773fb62848b88634ada49890082","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 10:01:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d344d017a2704a7783a8164ce524ac0e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1973a48b99b46cfa095cffb216e35a0 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4d61af64a1a9d0c28934d8bc16524487895471d0","revisions":{"4d61af64a1a9d0c28934d8bc16524487895471d0":{"kind":"REWORK","_number":1,"created":"2024-02-05 14:03:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/907742/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/42/907742/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/42/907742/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/42/907742/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/42/907742/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/42/907742/1"}}},"commit":{"parents":[{"commit":"e10099bd800c2ffd4f084b00d0d063103dc45fe0","subject":"Bump hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e10099bd800c2ffd4f084b00d0d063103dc45fe0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-05 14:02:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-05 14:02:51.000000000","tz":540},"subject":"Add upgrade job from 2023.1","message":"Add upgrade job from 2023.1\n\nThe 2024.1 release will support direct upgrade from 2023.1, so this\nscenario should be tested in CI.\n\nChange-Id: Idbe56fd0eaa2e4b04ce64051bd706ad7757cddf2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/4d61af64a1a9d0c28934d8bc16524487895471d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/4d61af64a1a9d0c28934d8bc16524487895471d0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
