)]}'
{"id":"openstack%2Fplacement~919579","triplet_id":"openstack%2Fplacement~master~I40410f637fcb1b05724a4ed7106e557c07006164","project":"openstack/placement","branch":"master","topic":"pre-commit","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2024-07-20 07:35:17.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-10-01 21:28:01.000000000","reason":"Change was submitted"},"34860":{"account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"last_update":"2024-10-01 21:28:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I40410f637fcb1b05724a4ed7106e557c07006164","subject":"Integrate pre-commit","status":"MERGED","created":"2024-05-14 14:13:21.000000000","updated":"2024-10-01 21:29:10.000000000","submitted":"2024-10-01 21:28:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"919579-pre-commit","meta_rev_id":"4b680b35d69a04d50e7b0e30ae58bc065bf15dd7","_number":919579,"virtual_id_number":919579,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-01 21:28:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2024-07-20 07:35:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-07-08 09:51:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":2,"date":"2024-10-01 20:10:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":1,"date":"2024-10-01 20:10:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-14 15:34:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-03 09:29:53.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"state":"CC"},{"updated":"2024-07-03 12:00:59.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2024-07-08 09:51:42.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"state":"REVIEWER"},{"updated":"2024-10-01 20:10:43.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"888542942e8444d2172f2027c94bc08e3e4471b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-14 14:13:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"455894fd1fe763e065dec560c81e6ad156e32c2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 15:34:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43bc37e289d54100b4a61b58a34dae79\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b7859bf5180a42a0b86a9b5617c31577 : SUCCESS in 5m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/e9f05a993945487899b57bdeea09c8aa : SUCCESS in 1h 07m 58s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/d05dc8f19c5b48419d4926b7bff8e151 : SUCCESS in 1h 19m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/daa57b7596b547ed91835f20deee3690 : SUCCESS in 1h 06m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e245086c34ee44fd8e9d29cb84d5624f : SUCCESS in 9m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fd980550bd443eb9f15746d7920e206 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b68a6f7749e04fd8b03bb95cb87b60da : SUCCESS in 6m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c9b6695bc6284b6ba515cbc0cbce69cf : SUCCESS in 4m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1cf661bc546643969b1ed2f7c774520c : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62136195463443baa452d5a6ef77a083 : SUCCESS in 8m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6db03f8c553946559fd853899c38cc8f : SUCCESS in 5m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/296415ba06464079b99f6563734fcdfc : SUCCESS in 7m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c102fe1ae8cd4a9887693e11be99ae1b : SUCCESS in 6m 10s\n- placement-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6a6850e4ccaa4d63b8b6a6c966c27e04 : SUCCESS in 6m 09s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f668d79ba3f242069f28a0925647bb02 : SUCCESS in 26m 22s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/ffd75a043cc145f2bc45e0f9e6cddddd : SUCCESS in 25m 30s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/a6421c2df78740c0bb6501c715f2080f : SUCCESS in 9m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80da5a7f4de04972b62c623c698eb597 : SUCCESS in 1h 05m 09s","accounts_in_message":[],"_revision_number":1},{"id":"7a9464026bc236ea948159f2cbe1e75df55867a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-06-13 09:54:19.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0e3c3d2f8f8b404b7fe430e3c5dc030c0e51bb4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 11:40:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/917a7a5a80814d61b2c8dbf1169ff0da\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f70fc52949904faaaaece3be4d441273 : SUCCESS in 4m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/18817b5cb71941499aa90808ee73fbe9 : SUCCESS in 1h 12m 46s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/ea64d84556534aee83447e94ba2d6426 : SUCCESS in 1h 40m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca55e392e40b4d9ca5a0137308b047fb : SUCCESS in 1h 07m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/495d6c2d26754a7885d0e6d0a462b708 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a51b8e93236748c3a46ec68cf00904f7 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c932b74f7ae74b3fbb773ef7ff5fc81a : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/696fc674272e4085928a9baaa815a73d : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ce464671a17b43d7b34438a271c3b6d7 : SUCCESS in 10m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2983e7d1ae5f42c39900eb1efdef1a7a : SUCCESS in 8m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5219622b01f4bb09becad518dd34701 : SUCCESS in 4m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6d97ca4d28104172a8231b3263810031 : SUCCESS in 5m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/36455fd0cc3e4a088b9bbeda8e1fdbab : SUCCESS in 5m 19s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b179a4dcfe08425895e14db551b71bab : SUCCESS in 21m 50s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/484831adc3574d64988e955cc3f288b0 : SUCCESS in 22m 36s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/88b3068aeed6476099d930b0ccd9db2f : SUCCESS in 11m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/40450f5820ac4bc08cca8a0f0be20710 : SUCCESS in 1h 10m 24s","accounts_in_message":[],"_revision_number":2},{"id":"7eaef115b7fa8852d56cbf1a2e940f8a0480691a","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-07-03 09:29:53.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b91d3b7ca78a786fdb6e191ef8d1cac37a5d4c63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-07-04 10:20:51.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"41bd4c3623f11bb3829f992db55623e69c982893","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 12:22:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49760297ae514ba8930f36105811693f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ece4d57fa5b54450b87705b8d676e966 : SUCCESS in 5m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/db5ba4f592e843a89aee25ad4e98dc6a : SUCCESS in 1h 01m 05s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/1f966aa95b1f4f71b3854506f0c1b983 : SUCCESS in 1h 37m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a129661595c1481996be21145f7eadfa : SUCCESS in 56m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ed88e11277444f79fc0853578344617 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/918981a063f94c71b2126b53da2aa72e : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/210e6a1525d24940b749ef5b816515cb : SUCCESS in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b4fc6c88cccf417cab4716f91b14ea3d : SUCCESS in 8m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e6be72c39f4c4d6294b9423b746fc7b5 : SUCCESS in 10m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d528cd83631e49bea51bf5e3a509008a : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c65447d2353498ba714ee1d11a88a2a : SUCCESS in 4m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/56947e7235f3453ea4473a756820094e : SUCCESS in 5m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ac4114465820448d840d0a22be20ea00 : SUCCESS in 7m 28s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e1992bc11da649ad97fd4e65009b4005 : SUCCESS in 19m 58s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/e236f442ab3640fd9b61446f199e948f : SUCCESS in 21m 29s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/50e17989f6ba49349fa188f0fc3153f0 : SUCCESS in 10m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef9ec349adeb4219a76b36eef58bdda8 : SUCCESS in 54m 30s","accounts_in_message":[],"_revision_number":3},{"id":"f09382ce70d31d0c8ed7fe00de4ea31564a67fcf","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-07-08 09:51:42.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6a8dab2ad8133428a18d1cb9db1acad485695603","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-07-20 07:35:17.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"efd59a31c8541c1342f5833d363266a66af4c9c9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-01 20:10:43.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e248fc25f303efa50e89b2a622e61bcac2849c88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 20:11:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cb34c47ceb4d864fa1f4a006cef62d6f903f7edc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 21:28:01.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0db85ffe44224f22bf28fdba2c6a47c0\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/515b539cdb2040659e2426a940e81cfb : SUCCESS in 5m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/efce6b47fd0f4093bb659a4ab690547e : SUCCESS in 46m 39s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/d57a115b6497494dad5e2fe890277687 : SUCCESS in 1h 16m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3ed4f1ce501441dbaf405a9643a1a75f : SUCCESS in 1h 02m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/08d88cebf3424394bb4e764fd55a83b9 : SUCCESS in 22m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e0776b659e94297b25e4a9ad4b6d3bf : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94616112bc714d34bab4dd05aceddf98 : SUCCESS in 3m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/01313a3de7eb441aa11e1eeffc0fe1bd : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34fd9d8a3fcc449487e1024e9057ac17 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3789d88e349146eeb22660b791cae6f0 : SUCCESS in 4m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/99bf1626198441309978a5f77adf189b : SUCCESS in 3m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fe7867113dae4945888ad1d71bf8366c : SUCCESS in 5m 39s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5cfa7f59a28140ecb6938f601e217dbb : SUCCESS in 19m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b37df20b85a34682acca5fd938722f4d : SUCCESS in 35m 12s","accounts_in_message":[],"_revision_number":3},{"id":"43e0e735c227b5da41dac31b9160425868d51b4e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 21:28:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"4b680b35d69a04d50e7b0e30ae58bc065bf15dd7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 21:29:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f358e48b0b14f26ae0bbeee51ea916c\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/ff86edb0f4694b149661c6f387c98dd7 : SUCCESS in 48s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1f65b49364f461eab6817e16a624a24 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3849bc4879b1462ab946326834307869 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"094867041d67098b69ab66ae604ea4059ae19e56","revisions":{"cfdaf9836aeeaf8b3d9770f3a121114a694a8c05":{"kind":"REWORK","_number":1,"created":"2024-05-14 14:13:21.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/79/919579/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/79/919579/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/79/919579/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/79/919579/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/79/919579/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/79/919579/1"}}},"commit":{"parents":[{"commit":"728abf5647a4344ab13151f930755845b7bbbdee","subject":"tox: Remove envdir overrides","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/728abf5647a4344ab13151f930755845b7bbbdee"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-14 13:29:44.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-14 13:32:29.000000000","tz":60},"subject":"Integrate pre-commit","message":"Integrate pre-commit\n\nJust as we\u0027ve done for Nova and many other projects.\n\nChange-Id: I40410f637fcb1b05724a4ed7106e557c07006164\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/cfdaf9836aeeaf8b3d9770f3a121114a694a8c05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/cfdaf9836aeeaf8b3d9770f3a121114a694a8c05"}]},"branch":"refs/heads/master"},"b0c4b4782de85b2fa7a10d076935e2e0f5f88a7a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-06-13 09:54:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/79/919579/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/79/919579/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/79/919579/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/79/919579/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/79/919579/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/79/919579/2"}}},"commit":{"parents":[{"commit":"51cbfeedba812132ca34839fd43405dbc2b88111","subject":"tox: Simplify functional testenv definitions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/51cbfeedba812132ca34839fd43405dbc2b88111"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-14 13:29:44.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-06-13 09:53:44.000000000","tz":60},"subject":"Integrate pre-commit","message":"Integrate pre-commit\n\nJust as we\u0027ve done for Nova and many other projects.\n\nChange-Id: I40410f637fcb1b05724a4ed7106e557c07006164\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b0c4b4782de85b2fa7a10d076935e2e0f5f88a7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b0c4b4782de85b2fa7a10d076935e2e0f5f88a7a"}]},"branch":"refs/heads/master"},"094867041d67098b69ab66ae604ea4059ae19e56":{"kind":"REWORK","_number":3,"created":"2024-07-04 10:20:51.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/79/919579/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/79/919579/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/79/919579/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/79/919579/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/79/919579/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/79/919579/3"}}},"commit":{"parents":[{"commit":"1a7a807592567bed34fe3576311b3289c30c6df9","subject":"tox: Simplify functional testenv definitions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/1a7a807592567bed34fe3576311b3289c30c6df9"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-14 13:29:44.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-07-04 10:13:23.000000000","tz":60},"subject":"Integrate pre-commit","message":"Integrate pre-commit\n\nJust as we\u0027ve done for Nova and many other projects.\n\nChange-Id: I40410f637fcb1b05724a4ed7106e557c07006164\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/094867041d67098b69ab66ae604ea4059ae19e56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/094867041d67098b69ab66ae604ea4059ae19e56"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
