)]}'
{"id":"openstack%2Fplacement~919580","triplet_id":"openstack%2Fplacement~master~I9d9d8b372b0a9c761b605188b3c641c9d1f326cb","project":"openstack/placement","branch":"master","topic":"pre-commit","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2024-07-20 07:38:22.000000000","reason":"removed on reply"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-10-01 21:30:05.000000000","reason":"Change was submitted"},"34860":{"account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"last_update":"2024-10-01 21:30:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9d9d8b372b0a9c761b605188b3c641c9d1f326cb","subject":"pre-commit: Add sphinx-lint","status":"MERGED","created":"2024-05-14 14:13:21.000000000","updated":"2024-10-01 21:31:51.000000000","submitted":"2024-10-01 21:30:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919580-pre-commit","meta_rev_id":"29865f22a2956b04f579b1070710847f1f587f57","_number":919580,"virtual_id_number":919580,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-01 21:30:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2024-07-20 07:38:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2024-10-01 20:11:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2024-10-01 20:11:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-14 15:52:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-03 10:07:28.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"state":"REVIEWER"},{"updated":"2024-07-03 12:01:08.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2024-10-01 20:11:46.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"d99865298ab58ac262b8d4c014720460ed10c126","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-14 14:13:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ecfd95114fbf93705d11d54b54e6d7a3307d2c5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 15:52:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66cec7d6e9c54473b6948fc7587359bf\n\n- grenade https://zuul.opendev.org/t/openstack/build/b00617b4a0254887b034dcc50e04050b : SUCCESS in 1h 19m 40s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/d0c8267145a441f9b158da4c9c998f6a : SUCCESS in 1h 35m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1093220069934e74a13277aa2158b2e1 : SUCCESS in 1h 07m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c15e578bb3e404fbc46e21f877a49d0 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5de150a38ca45a09c7bf417f1e093c9 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41788c3511fc482182e4ce9c0edf2f14 : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e59a3fff13114bc3a906a92e7d689305 : SUCCESS in 4m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f3cd7fd69241403f9506b11c61fc6ac9 : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cc59cbbf5474da09ea3765fbcba6191 : SUCCESS in 6m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/68f8f464389a4abeb21d1c8922ad6b0b : SUCCESS in 5m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/736997f0621547c69423cea85ad3327d : SUCCESS in 6m 51s\n- placement-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d059b8da6161441291da52fe177d7a56 : SUCCESS in 6m 08s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b38ce808ec744ced843702283e902241 : SUCCESS in 20m 28s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/3d946758521c4e34ac576cbd4f9e0f98 : SUCCESS in 34m 29s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/a5dc5f1a12b94772b575769601127111 : SUCCESS in 9m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1d6a1a56a5442f5925b8f1aa5b9f601 : SUCCESS in 1h 02m 31s","accounts_in_message":[],"_revision_number":1},{"id":"5e01ec6ad2a2b76cde639a79abb19d4ff8a45443","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-06-13 09:54:19.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"c1b3a6a1e7874eede9c12ac6513c446f72bbff35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 11:41:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87a59f1d270c4c1f8860648f3d12e0a3\n\n- grenade https://zuul.opendev.org/t/openstack/build/628c2a8bf96f40a9bb6f83ae95bdb444 : SUCCESS in 57m 59s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/591834a5549a4c84854c0ef757f07926 : SUCCESS in 1h 45m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c469e78f1da8422dbee0070365d944dd : SUCCESS in 1h 03m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f2ab4bac31e47b9a4b6d6f555abb1bb : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/773bc559cb3a4beb936a05bc600ec5cb : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9cc0679fcb54a188db472c266456e36 : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/85f099682cb74408b02ba120cbe9874e : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/de1c9842c4b9473dab47c0e88f679424 : SUCCESS in 10m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1b518de1c054d95aeb53ee0b7446e16 : SUCCESS in 7m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d26641d58c164853afc4f1612ef630f9 : SUCCESS in 5m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e58689597f904d948110d670e7512158 : SUCCESS in 6m 25s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2e03a610c0194bde88ed0a0c70efd2f7 : SUCCESS in 21m 40s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/852d371016d542a3a5eb6ee0eca84de1 : SUCCESS in 29m 07s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/271ba0769e664fa3bfc19be25fd4cdaa : SUCCESS in 9m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/33afd8fb8e2d4545b12cdfa2046ce05e : SUCCESS in 1h 02m 22s","accounts_in_message":[],"_revision_number":2},{"id":"8f1dcc94369777a4c3e19023d0222016c3d06f87","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-07-03 10:07:28.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8c1c81f15292452cdee19255fac19047fe590c23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-07-04 10:20:51.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3728b7fa50ab9772e52076a1fe578b8d16726f4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 12:13:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a21daeb50384301ba99bc534c694b06\n\n- grenade https://zuul.opendev.org/t/openstack/build/d4384ac23e8946b9b9368f19ba01be42 : SUCCESS in 1h 00m 40s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/c9eab760cb4344c78a286acb6389841d : SUCCESS in 1h 33m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee020a6dbe9d46fba8f1a3ab14ec6b3f : SUCCESS in 1h 03m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7c740d2318f43efacf13783a8c83bdf : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a580f94fefa74f5781abb487ee74586a : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00f25e9c00e840018b53fbe1bf3daa99 : SUCCESS in 8m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/27ebadb097be4cbabfc92851deecc8db : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bb29d7e8327644fbbfc95367a734cd2b : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/213b52eee2864069a8ca8e126556ecad : SUCCESS in 6m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4c0456436818453baad4ae510f10e0c7 : SUCCESS in 6m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5cc3390269554446bd0272f9b035b342 : SUCCESS in 5m 16s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ad4e22e1e5ad4d6b8ff1d873c9d35c01 : SUCCESS in 18m 42s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/2cb63ea33f34484e97ac91bcdacd9480 : SUCCESS in 27m 01s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/8feaeded685042c1b9d06c2263437119 : SUCCESS in 12m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b687b3b6a7d54f52aa6a12c966cc89c0 : SUCCESS in 56m 18s","accounts_in_message":[],"_revision_number":3},{"id":"d831180d002d7f253534cb35b203a0ee3ad8e564","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-07-20 07:38:22.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8be4fad4208d0b768b91d2a6c368d515fddc4e79","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-01 20:11:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5d9a3e631943f0adb529c521b8d047c3f60e19d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 20:12:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a88d6edf4607a9ad5e28cc6cd60188b6f40fe15d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 21:30:05.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24a693f9d0c741fda3517095d5158227\n\n- grenade https://zuul.opendev.org/t/openstack/build/697e9156184542b3b2ca0824303d23ae : SUCCESS in 46m 00s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/d03a1f58d22b4b81a54eb2829987f581 : SUCCESS in 1h 13m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b782a57ae0544eb5a2a561f638f523e1 : SUCCESS in 53m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/81d48560aa754bb18b937de060bb1a79 : SUCCESS in 33m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89c6a5258b784e7a9ac7cf5ddc13b02d : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ab120db531c46a6b015828714e9b0ed : SUCCESS in 5m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cde5773f5fd74deeb3c2f1b7cc82941a : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2a11821362a4132b10676c11143788f : SUCCESS in 6m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/65cb682a584f4484bf755dc5fd4171ee : SUCCESS in 5m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/42a528fe04274892993df1a4c831bd08 : SUCCESS in 6m 26s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/53715d77c0f74ed0b13f59625b28b775 : SUCCESS in 16m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81e1cafe67054027ace9dbab86d0dfb5 : SUCCESS in 52m 28s","accounts_in_message":[],"_revision_number":3},{"id":"f08d0e6d916a4d3f7b83d6aece0c306a5c7b00b3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 21:30:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"29865f22a2956b04f579b1070710847f1f587f57","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 21:31:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98f69e1605254f0aa85f2d6e86617407\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19485ff2b59742d7a5a2942a32d40f4c : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0736c6105699941d1e7493477fc21e5eef6ddccd","revisions":{"4c983b37b5b33d2076f571b5aa9b3f3e0252be86":{"kind":"REWORK","_number":1,"created":"2024-05-14 14:13:21.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/80/919580/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/80/919580/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/80/919580/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/80/919580/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/80/919580/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/80/919580/1"}}},"commit":{"parents":[{"commit":"cfdaf9836aeeaf8b3d9770f3a121114a694a8c05","subject":"Integrate pre-commit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/cfdaf9836aeeaf8b3d9770f3a121114a694a8c05"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-14 13:32:43.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-14 13:41:07.000000000","tz":60},"subject":"pre-commit: Add sphinx-lint","message":"pre-commit: Add sphinx-lint\n\nChange-Id: I9d9d8b372b0a9c761b605188b3c641c9d1f326cb\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/4c983b37b5b33d2076f571b5aa9b3f3e0252be86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/4c983b37b5b33d2076f571b5aa9b3f3e0252be86"}]},"branch":"refs/heads/master"},"6aaa6e8f07207e238787104c2d3d5e2e584e7cde":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-06-13 09:54:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/80/919580/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/80/919580/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/80/919580/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/80/919580/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/80/919580/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/80/919580/2"}}},"commit":{"parents":[{"commit":"b0c4b4782de85b2fa7a10d076935e2e0f5f88a7a","subject":"Integrate pre-commit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b0c4b4782de85b2fa7a10d076935e2e0f5f88a7a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-14 13:32:43.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-06-13 09:53:44.000000000","tz":60},"subject":"pre-commit: Add sphinx-lint","message":"pre-commit: Add sphinx-lint\n\nChange-Id: I9d9d8b372b0a9c761b605188b3c641c9d1f326cb\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/6aaa6e8f07207e238787104c2d3d5e2e584e7cde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/6aaa6e8f07207e238787104c2d3d5e2e584e7cde"}]},"branch":"refs/heads/master"},"0736c6105699941d1e7493477fc21e5eef6ddccd":{"kind":"REWORK","_number":3,"created":"2024-07-04 10:20:51.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/80/919580/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/80/919580/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/80/919580/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/80/919580/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/80/919580/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/80/919580/3"}}},"commit":{"parents":[{"commit":"aa520fec9c58846f92076bcbe7fa76de02abb643","subject":"pre-commit: Add autopep8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/aa520fec9c58846f92076bcbe7fa76de02abb643"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-14 13:32:43.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-07-04 10:13:36.000000000","tz":60},"subject":"pre-commit: Add sphinx-lint","message":"pre-commit: Add sphinx-lint\n\nChange-Id: I9d9d8b372b0a9c761b605188b3c641c9d1f326cb\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/0736c6105699941d1e7493477fc21e5eef6ddccd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/0736c6105699941d1e7493477fc21e5eef6ddccd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
