)]}'
{"id":"openstack%2Fplacement~919582","triplet_id":"openstack%2Fplacement~master~Ib7ea7fa00f8a4cc97f20e01fbd0357f737a6c224","project":"openstack/placement","branch":"master","topic":"pyproject","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-11-25 14:02:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib7ea7fa00f8a4cc97f20e01fbd0357f737a6c224","subject":"setup: Remove pbr\u0027s wsgi_scripts","status":"MERGED","created":"2024-05-14 14:22:41.000000000","updated":"2025-11-25 14:03:21.000000000","submitted":"2025-11-25 14:02:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919582-pyproject","meta_rev_id":"775fa417c8f7a04bf5146efdfbedaaaebc7976d7","_number":919582,"virtual_id_number":919582,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-25 14:02:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2025-11-25 11:45:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2025-11-25 12:06:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2025-11-25 12:06:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-14 15:55:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-22 11:01:06.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-11-25 12:06:16.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"1fbfd5648bf2bc25c6d7fe10375640504aa17c8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-14 14:22:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8425a50753fc546b643954ecd8bb456c9117f0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 15:55:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13b376656b0746158318b70e2b5df3c0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2be7dc0a4937497e92278c1f34f207af : SUCCESS in 2m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/af06d64b48b84bd59680e955a8a02a39 : SUCCESS in 58m 45s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/2c64153b82c84bbfbb4ec75a8a2d1e21 : SUCCESS in 1h 28m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a16c1636fd82400099058639f5e8f823 : SUCCESS in 1h 04m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44f3708c37a946a98dbc231941436033 : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8929d8904d64eaf845c10d7c18ac682 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9deaa20c51b04b71b9684113653a5c5d : SUCCESS in 5m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/85ef1b1b16cd44adaf9e231160687495 : SUCCESS in 4m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c1b254aab5c34cad830313d38e3dbc8f : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21113e4e71b04728a6163f96869f1ab3 : SUCCESS in 8m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fadb9e13c8924901b5d3919b199d2c1b : SUCCESS in 5m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/eaac2e12cbe74aeaa117adb2631ce96d : SUCCESS in 5m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1955c518332549c09f5efd1125987ef4 : SUCCESS in 5m 32s\n- placement-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2dd55feb856e42149e596e7ba5bf5191 : SUCCESS in 6m 52s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7d12f2d0847e410592957e3b41683e01 : SUCCESS in 20m 31s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/85cb8dbcea7446e885200523046bf104 : FAILURE in 4m 54s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/fd3b8fbb83a945a48087c7b86047d2be : FAILURE in 3m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2febd3302d864604a79f2a327741cfa7 : SUCCESS in 1h 06m 05s","accounts_in_message":[],"_revision_number":1},{"id":"138b92b9803ed86201c7c9da2d65926b7361f302","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-07-22 11:01:06.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"910d395f636029f34dc8cbb72b3569d61064ce35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-06 14:54:30.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"4560da34a154e315c25fe464149927153618eaa6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-06 14:55:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"61d7d1db5b204c7457b691e61c185953782c433e","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-05-06 14:56:08.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"5324afb0392241d8d1ff44628d2969737a95a2c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 16:44:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a4886f6cc004400bf1f01a9fcc2d93d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/eaf6ec74294341a490aea72bf249e06f : SUCCESS in 2m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/c4c22fc62c5a48bdb6e1d6ce1d9ab569 : SUCCESS in 1h 04m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dc03d5e1a424490aacafb2d31ea39876 : SUCCESS in 47m 30s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/e7ab5c2afa48446d92c0e2e35d8534a1 : SUCCESS in 1h 33m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/40516b6ab7a540e1bf5fba72fd2fb1d2 : SUCCESS in 1h 04m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e804f7aad7543a5bc9ad4a14a870b74 : SUCCESS in 11m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92d60fe85c55437f89de41e8c3fd8abe : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eec552dda1f043eab4c07b4dc16ba9f1 : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5cbee79f97f5460b864aefbd654d872b : SUCCESS in 7m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/54e126a798244292a190f3de65f7a90a : SUCCESS in 9m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6707c9a11e8d42f9ac6e15f3df394600 : SUCCESS in 10m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1adec4368bca44dc9f0f9ee40e6a9b80 : SUCCESS in 5m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/08c56e2492c5446d849058c88997ee2b : SUCCESS in 8m 43s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/921c4c4e202d44d68f075764a069c954 : SUCCESS in 11m 13s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fa2524444dc64c0da748283684b2ef07 : SUCCESS in 19m 43s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/48e98140e36e42d29d7aa06ec8db86d0 : FAILURE in 10m 20s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/2b083b83029d45a0841a50b9ab97c897 : FAILURE in 4m 45s (non-voting)\n- tempest-integrated-placement-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71ad9a92a73d4c55ade338d6d059f777 : SUCCESS in 1h 28m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/10146e965acb4f0eb7369a9ada82a933 : SUCCESS in 57m 26s","accounts_in_message":[],"_revision_number":2},{"id":"61efbd26b9084311ca5857896b2c41e58fca2bf5","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-05-06 18:23:04.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7bdf6dbaa1f0d709ff2d56948b3c2dc79af010c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-18 19:00:44.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"9fc8c45754d32801b0fa2b8fef5d7534eb34ba94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 20:54:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d32cde83eb54f159ce32f8827b1a1d8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d729620d929a4b06b5acdbda72a990a1 : SUCCESS in 2m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/18c2cdacfe6246e997b8f578307d4924 : SUCCESS in 41m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/11d26e32728a4b1e8e49c826845f6079 : SUCCESS in 48m 54s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/61e97b0b969c434dbc4d82f5770120c2 : SUCCESS in 1h 48m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/511701df00a94d818f7dfaaaf60eac0e : SUCCESS in 1h 06m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5248442ff55540f4a081d647b74f543a : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/369939c2fae847f498170c0cea4b4b43 : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/398029e9e25b4c7ba94bd01f9147c9d2 : SUCCESS in 5m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8fb02109ec4044cc997d9c6b35e460c8 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ee72aa0c68f4a468cbdd5cf6c6a22e8 : SUCCESS in 7m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcd5d6fa90f9475e83ba8de7a1d5e133 : SUCCESS in 4m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5eb87961cfe847fb99b854167b9fae2b : SUCCESS in 7m 03s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/990ebfa5f3684fecb1e47875c5923cbb : SUCCESS in 8m 14s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a581af9aa6444ee1a573bd5445c8aa57 : SUCCESS in 23m 58s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/af1eac9827304bc4a54b90602cc4727c : FAILURE in 10m 15s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/ce06627a3def447bb2796727a1601fae : FAILURE in 3m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73891ba48c844ce28f01b16fc778664f : SUCCESS in 45m 54s","accounts_in_message":[],"_revision_number":3},{"id":"a65ca9bdb1871af580e7930539f23a413649a14b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-11-25 11:45:59.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b0b121fc7811d09dae1aeda13d415feb748e4815","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-11-25 12:06:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c8bf8b65b4ab966a5ab3886f2b8380d317631edc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 12:06:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"208261dbb5dd9eba8af65d88073c9352645be190","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 14:02:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91c65c2aad01434ab5a1418249e4e77b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9fcb6cbb5f86495db88f3cee8e9acedf : SUCCESS in 3m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/557c3dffd1ca48d4a3ca01f376a0557a : SUCCESS in 29m 29s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/e6abfa99398c4a34b8a66c5b4cd7c97d : SUCCESS in 1h 08m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3808b21856fe46b6aa5e89a2702e73df : SUCCESS in 1h 11m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b797f478d724d2caa656047f025b996 : SUCCESS in 36m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d92cf8bb8fa449b392f1a0f580024007 : SUCCESS in 5m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/843f6646cffd49afad1d3ef316c3dfc5 : SUCCESS in 5m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f2b3216bd37c412ba989ce62d66eee0d : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e407dc34064549a5b08c4c27ab2118aa : SUCCESS in 10m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51f94e34d906496eb7f594723596a511 : SUCCESS in 5m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/13fc15da28cc4bb994d6b5cb3109f918 : SUCCESS in 4m 05s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4fa4122940954f0bac2737ce32795af4 : SUCCESS in 9m 05s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ed16ac12eca6413392aa775619a679b9 : SUCCESS in 21m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae9d5ba479ca494888fad0f391459c13 : SUCCESS in 31m 34s","accounts_in_message":[],"_revision_number":3},{"id":"df56fad1394f6cce6aac07dbaa4fe6a86469295d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 14:02:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"775fa417c8f7a04bf5146efdfbedaaaebc7976d7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 14:03:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e921632596745dca2c8ad327519e361\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5675e5dd28734eb0989052fba2d36326 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ee6e3e654fb419f865cb996b5fd3cbb : SUCCESS in 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"39f4895d9c5361ea529180d68f8bf92fcba1fd2a","revisions":{"f57e4ca39b987c86f579b48227125e68fe39f9e8":{"kind":"REWORK","_number":1,"created":"2024-05-14 14:22:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/919582/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/82/919582/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/82/919582/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/82/919582/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/82/919582/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/82/919582/1"}}},"commit":{"parents":[{"commit":"5a17e40555d6ea59216de7af1c271f6bed8ac48f","subject":"Add placement.wsgi.api module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/5a17e40555d6ea59216de7af1c271f6bed8ac48f"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-14 14:21:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-14 14:22:25.000000000","tz":60},"subject":"setup: Remove pbr\u0027s wsgi_scripts","message":"setup: Remove pbr\u0027s wsgi_scripts\n\nThis is technical dead end and not something we\u0027re going to be able to\nsupport long-term in pbr. We need to push users away from this.\n\nChange-Id: Ib7ea7fa00f8a4cc97f20e01fbd0357f737a6c224\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nDepends-on: https://review.opendev.org/c/openstack/devstack/+/919581\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f57e4ca39b987c86f579b48227125e68fe39f9e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f57e4ca39b987c86f579b48227125e68fe39f9e8"}]},"branch":"refs/heads/master"},"ffe88b8d906b7c69f73edb4839deafd6fb077213":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-06 14:54:30.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/919582/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/82/919582/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/82/919582/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/82/919582/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/82/919582/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/82/919582/2"}}},"commit":{"parents":[{"commit":"124ba4514122c30456012644d6a90dc9739135d4","subject":"Merge \"Add pyproject.toml to support pip 23.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/124ba4514122c30456012644d6a90dc9739135d4"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-14 14:21:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-06 14:54:30.000000000","tz":0},"subject":"setup: Remove pbr\u0027s wsgi_scripts","message":"setup: Remove pbr\u0027s wsgi_scripts\n\nThis is technical dead end and not something we\u0027re going to be able to\nsupport long-term in pbr. We need to push users away from this.\n\nChange-Id: Ib7ea7fa00f8a4cc97f20e01fbd0357f737a6c224\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nDepends-on: https://review.opendev.org/c/openstack/devstack/+/919581\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ffe88b8d906b7c69f73edb4839deafd6fb077213"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ffe88b8d906b7c69f73edb4839deafd6fb077213"}]},"branch":"refs/heads/master","description":"Rebase"},"39f4895d9c5361ea529180d68f8bf92fcba1fd2a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-11-18 19:00:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/919582/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/82/919582/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/82/919582/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/82/919582/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/82/919582/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/82/919582/3"}}},"commit":{"parents":[{"commit":"f465838dbceb4f47696300eef0cf7b9fdde7d90d","subject":"Make sure [cors] allowed_origin accepts a list value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f465838dbceb4f47696300eef0cf7b9fdde7d90d"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-14 14:21:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-18 18:56:53.000000000","tz":0},"subject":"setup: Remove pbr\u0027s wsgi_scripts","message":"setup: Remove pbr\u0027s wsgi_scripts\n\nThis is technical dead end and not something we\u0027re going to be able to\nsupport long-term in pbr. We need to push users away from this.\n\nChange-Id: Ib7ea7fa00f8a4cc97f20e01fbd0357f737a6c224\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nDepends-on: https://review.opendev.org/c/openstack/devstack/+/919581\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/39f4895d9c5361ea529180d68f8bf92fcba1fd2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/39f4895d9c5361ea529180d68f8bf92fcba1fd2a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
