)]}'
{"id":"openstack%2Fplacement~926743","triplet_id":"openstack%2Fplacement~master~I41392ad959fa5c8bbcdeb5f0b7ea9819dea7690b","project":"openstack/placement","branch":"master","topic":"bug/2039560","attention_set":{},"removed_from_attention_set":{"25733":{"account":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"last_update":"2024-10-07 12:48:56.000000000","reason":"\u003cGERRIT_ACCOUNT_25733\u003e replied on the change","reason_account":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"}},"9542":{"account":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"last_update":"2024-11-08 19:21:14.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I41392ad959fa5c8bbcdeb5f0b7ea9819dea7690b","subject":"DNM -- testing fix in external PR","status":"ABANDONED","created":"2024-08-21 11:55:19.000000000","updated":"2024-11-08 19:21:14.000000000","total_comment_count":9,"unresolved_comment_count":5,"has_review_started":true,"meta_rev_id":"a4e6e20188247141e1779d063e8cd5f58c015403","_number":926743,"virtual_id_number":926743,"owner":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-21 11:55:19.000000000","updated_by":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"reviewer":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"state":"CC"},{"updated":"2024-08-21 13:50:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fb8f26d4eb040376c1c5f8a06ddffa988de65103","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"date":"2024-08-21 11:55:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc9c51944a1e329b0219289c1fb3ee45560ea0fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"date":"2024-08-21 12:18:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"591958dd8c1b39e989b2d171f2a769b7a872979b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 13:50:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/be0598448c714c4a8321ffd9064d6553\n\n- grenade https://zuul.opendev.org/t/openstack/build/c8fc7698f8524f038f11fe592360d351 : SUCCESS in 1h 03m 22s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/87d28bf3b28f4280887460cb0d0747f6 : SUCCESS in 1h 27m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47cedd806e134aad9bf7971e13673cee : SUCCESS in 57m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f50a031de03b42199edb5154b3e332c9 : FAILURE in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0808a2f8a6e14980ad586d0607bdde7a : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a8bd1dc93cd4bbc87109fca8f5d9f5f : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/32182f9c0c7f4bb5b8d68a0a256fab59 : SUCCESS in 4m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/07888c21f5784ac99dae57943492049c : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8a042565be0439bb1aad2657f8d6790 : SUCCESS in 7m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5df29b1fbe1c4d5585316e2347f565ae : FAILURE in 6m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/101daa48c45e4c49976d9843009d915b : FAILURE in 6m 10s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6ed066e20358405cbfc72ae9e79c1db3 : FAILURE in 23m 52s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/e2ca6cbde3294a6c87ceecb815e6f341 : SUCCESS in 27m 08s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/91a84cdc00ec4aca8a4076baf35edb8b : SUCCESS in 10m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b5e96cfadb9a453d9984fd15d31a238d : SUCCESS in 1h 01m 38s","accounts_in_message":[],"_revision_number":2},{"id":"3f66847346d8ef2711a78e24089e4291e64783cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"date":"2024-08-21 13:52:34.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"e1662e8ab54a1055280fc4dba791260d922fee88","author":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"date":"2024-08-21 14:22:12.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1162a3ff45310119176c69ff2cdd90a97f758504","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 15:58:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/34604affc0284cb0b73663f05507196e\n\n- grenade https://zuul.opendev.org/t/openstack/build/ee556f42e3474e3dbcb42e6fe4e69e76 : SUCCESS in 1h 08m 43s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/d548fceec2ae419e8a53d2034077afeb : TIMED_OUT in 2h 04m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1bca86b51b3b4c0990fb45bf275d7004 : SUCCESS in 57m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/199757635dd14166baf41ce74c2887b4 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a171f601b5164f3f863ac35857f29e09 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7634bb1f39c1427cb8bf20e2ff83314f : SUCCESS in 6m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/27f7b3672f484227a33975a9323f0d49 : SUCCESS in 3m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/012de895567c4566b8936ce86a1867ee : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bc75dcee77a43dc84ac5b485e5307e0 : SUCCESS in 7m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4a4d6f8e40014c938042062418d79479 : SUCCESS in 5m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2abf7f171c2e416fbe233a64c5b7592d : SUCCESS in 5m 09s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4b01aaaf05d24ba29345f238d16ebc4b : SUCCESS in 21m 45s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/a2ca4f7b0415480485eef0ee3d4c97b7 : SUCCESS in 31m 14s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/cb60cea0a6d94c5d9d094093aac85750 : SUCCESS in 11m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f367f625595748cea572a52969d42355 : SUCCESS in 58m 51s","accounts_in_message":[],"_revision_number":3},{"id":"05bb2cc6b432000f3999e43496cccdade3cbb094","author":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"date":"2024-08-21 16:06:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0e044f816ee85de79970057d6b8311f17d460319","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 17:42:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b907db703ae9427da99772b59b5596be\n\n- grenade https://zuul.opendev.org/t/openstack/build/2e78171260ae4d61bb9ff6b9bcbe0772 : SUCCESS in 1h 01m 29s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/52908d49077c482ab3efc0b419187aaf : SUCCESS in 1h 35m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c04e46f4caf41c2a4a2b9ba57f4678e : SUCCESS in 1h 09m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e6b7444cecb47719f604e2928567770 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cb07aa7aca64b23af77098e85b16435 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9968f2848d0d465aabeb59c0af7fedfd : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d4d5a379b11c42dba34cb650820f1ca0 : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0dfed824c1c04b699a2c6352b964869c : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a4a103f819c486c95fe517d25f61a33 : SUCCESS in 7m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7aa0d058c81f4c988bdf6bcd0e82e064 : SUCCESS in 4m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/075c3e54f2b44406a359cd1f457a406b : SUCCESS in 5m 13s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/251ce319ad5744c6b98bdd2bafb12679 : SUCCESS in 18m 20s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/ec061840089140bb921e845774ce6fc1 : SUCCESS in 20m 49s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/85e417b8d2a347dea89afdadfb0d76e0 : FAILURE in 2m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2dff397049534963a7082864ee99e482 : SUCCESS in 54m 14s","accounts_in_message":[],"_revision_number":3},{"id":"dd43a8cee1ea14cb52bc2a173b6cca5f6a5c5e48","author":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"date":"2024-10-04 16:52:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f54c3920d7d069afbdd785808757d7138549f0bb","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2024-10-07 10:44:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1ca5f64eaf388521c5a131b554d0d3ae98ff2bdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"date":"2024-10-07 11:00:55.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"dab074474909f17de60bbea06c6694da1547c91a","author":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"date":"2024-10-07 11:21:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"03ce0af57796dff0d6907b3eb9d882d8c8e9a8b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 12:19:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/366ccdd933204939961c58717150630a\n\n- grenade https://zuul.opendev.org/t/openstack/build/a90a4db5e508485ea16fbbe583178883 : SUCCESS in 52m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0ddd0bf6c0f64bd4b7aae61038ed628d : SUCCESS in 56m 07s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/f44b43fdf6224552b811eaec5944b4f7 : SUCCESS in 1h 16m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5292bd37a7f4196bb002ca8c4eeec98 : FAILURE in 38m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/064aa6d59c6c4941b01197025800a719 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6e8345c688c4a8a95bb484393018695 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4ae2bf70b2044ca9ebf4e4fc6086cd8 : SUCCESS in 6m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f5df79ca8be84211b3be925798a71622 : SUCCESS in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d6c106694c24562adea21d58d6f0017 : SUCCESS in 7m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/76647d4b5b6a49fa9faa1a76d2855615 : SUCCESS in 6m 27s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c65a08daed6141e2a6ee9dda71bb36f7 : SUCCESS in 8m 25s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1f9a7651d4c04e748497f5a7429378be : SUCCESS in 18m 44s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/2695fd88f5ad4a3ea4ea0feb99e5838a : SUCCESS in 18m 31s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/44a0261c63f045e4b65f6ecd293e2235 : FAILURE in 3m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/45b8ea51ad8a4539904c6c0079c90bde : SUCCESS in 1h 07m 55s","accounts_in_message":[],"_revision_number":4},{"id":"4a80cbebd86bfd87bf0a9ed399650515ef9d3ac6","author":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"date":"2024-10-07 12:48:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e25d764698bf2bb4140baa7440e70b7860f9010e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 14:09:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27574039fff6465599b07cb90e220638\n\n- grenade https://zuul.opendev.org/t/openstack/build/d5e8e305668440428c79d2b48378eabe : SUCCESS in 39m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c669919ae6a342979f1c571cf74f0664 : SUCCESS in 1h 00m 56s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/106fd0f5f90e4e32adf99e16c98b3bc7 : SUCCESS in 1h 19m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2bc644549877455da5680a29c8be21ee : SUCCESS in 54m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdfe45af26e84e43a4c49e124a816f49 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abe3da303d834cb7a548170363e7922b : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d3c6dbf49d9498fb3cd6007e2e56453 : SUCCESS in 5m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/08a5cf52184c4d7d9f04e611b03df72f : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3649d057a9d4731ac7f6eae2eb1654c : SUCCESS in 7m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/954895131dc9416395a053e0b292b289 : SUCCESS in 5m 42s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/56c6cc1731dd422597b15eacc3f7f73d : SUCCESS in 8m 31s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a4240417ea904338a973587391fe40d7 : SUCCESS in 18m 45s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/f54821e3432f4d50be9084bdf594038c : SUCCESS in 15m 49s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/68b804dd85ee4a36b8fa4c0c562dcaf4 : FAILURE in 3m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/180c821fe5654ec8962b24a041a40e1e : SUCCESS in 59m 02s","accounts_in_message":[],"_revision_number":4},{"id":"fffbaacf67d51fd2c9a1d66a73fc3736ea699de7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"date":"2024-11-08 18:58:49.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"a4e6e20188247141e1779d063e8cd5f58c015403","tag":"autogenerated:gerrit:abandon","author":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"date":"2024-11-08 19:21:14.000000000","message":"Abandoned\n\nAbandoned in favor of https://review.opendev.org/c/openstack/placement/+/934451","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"9b9007afa395241e4eca80b9728c85b108f89a4e","revisions":{"aa94fb0faaa5a1ca7b4675540310e1b704b9178e":{"kind":"REWORK","_number":1,"created":"2024-08-21 11:55:19.000000000","uploader":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"ref":"refs/changes/43/926743/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/43/926743/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/43/926743/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/43/926743/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/43/926743/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/43/926743/1"}}},"commit":{"parents":[{"commit":"ce3dd3839ba4df38c6ca5a1f3396bd2e3fad5bfa","subject":"Merge \"Changed /tmp/migrate-db.rc to /root/migrate-db.rc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ce3dd3839ba4df38c6ca5a1f3396bd2e3fad5bfa"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2023-10-18 15:30:26.000000000","tz":0},"committer":{"name":"Jan Gutter","email":"jan.gutter@workday.com","date":"2024-08-21 11:55:17.000000000","tz":60},"subject":"DNM -- testing fix in external PR","message":"DNM -- testing fix in external PR\n\nThis is testing the proposed fix (code only) in\nhttps://github.com/sapcc/placement/pull/5/files to see which\ntests (if any) are tripped.\n\n( this is a slightly modified version of\nhttps://review.opendev.org/c/openstack/placement/+/898808 )\n\nChange-Id: I41392ad959fa5c8bbcdeb5f0b7ea9819dea7690b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/aa94fb0faaa5a1ca7b4675540310e1b704b9178e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/aa94fb0faaa5a1ca7b4675540310e1b704b9178e"}]},"branch":"refs/heads/master"},"2497f1d079cefbe84f919f7abef3b32780350253":{"kind":"REWORK","_number":2,"created":"2024-08-21 12:18:14.000000000","uploader":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"ref":"refs/changes/43/926743/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/43/926743/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/43/926743/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/43/926743/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/43/926743/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/43/926743/2"}}},"commit":{"parents":[{"commit":"ce3dd3839ba4df38c6ca5a1f3396bd2e3fad5bfa","subject":"Merge \"Changed /tmp/migrate-db.rc to /root/migrate-db.rc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ce3dd3839ba4df38c6ca5a1f3396bd2e3fad5bfa"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2023-10-18 15:30:26.000000000","tz":0},"committer":{"name":"Jan Gutter","email":"jan.gutter@workday.com","date":"2024-08-21 12:18:06.000000000","tz":60},"subject":"DNM -- testing fix in external PR","message":"DNM -- testing fix in external PR\n\nThis is testing the proposed fix (code only) in\nhttps://github.com/sapcc/placement/pull/5/files to see which\ntests (if any) are tripped.\n\n( this is a slightly modified version of\nhttps://review.opendev.org/c/openstack/placement/+/898808 )\n\nChange-Id: I41392ad959fa5c8bbcdeb5f0b7ea9819dea7690b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/2497f1d079cefbe84f919f7abef3b32780350253"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/2497f1d079cefbe84f919f7abef3b32780350253"}]},"branch":"refs/heads/master"},"936f53ee84264ae5947c97ef0606db2e33020376":{"kind":"REWORK","_number":3,"created":"2024-08-21 13:52:34.000000000","uploader":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"ref":"refs/changes/43/926743/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/43/926743/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/43/926743/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/43/926743/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/43/926743/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/43/926743/3"}}},"commit":{"parents":[{"commit":"ce3dd3839ba4df38c6ca5a1f3396bd2e3fad5bfa","subject":"Merge \"Changed /tmp/migrate-db.rc to /root/migrate-db.rc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ce3dd3839ba4df38c6ca5a1f3396bd2e3fad5bfa"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2023-10-18 15:30:26.000000000","tz":0},"committer":{"name":"Jan Gutter","email":"jan.gutter@workday.com","date":"2024-08-21 13:51:18.000000000","tz":60},"subject":"DNM -- testing fix in external PR","message":"DNM -- testing fix in external PR\n\nThis is testing the proposed fix (code only) in\nhttps://github.com/sapcc/placement/pull/5/files to see which\ntests (if any) are tripped.\n\n( this is a slightly modified version of\nhttps://review.opendev.org/c/openstack/placement/+/898808 )\n\n- adding in the test fixture update from:\n  https://github.com/sapcc/placement/pull/5/commits/61164db5e07b7bc4146da63dc3051220752ccb52\n\nChange-Id: I41392ad959fa5c8bbcdeb5f0b7ea9819dea7690b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/936f53ee84264ae5947c97ef0606db2e33020376"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/936f53ee84264ae5947c97ef0606db2e33020376"}]},"branch":"refs/heads/master"},"4ecb80429f802ecbbeb0ca25f061597bc4d63b9c":{"kind":"REWORK","_number":4,"created":"2024-10-07 11:00:55.000000000","uploader":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"ref":"refs/changes/43/926743/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/43/926743/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/43/926743/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/43/926743/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/43/926743/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/43/926743/4"}}},"commit":{"parents":[{"commit":"ce3dd3839ba4df38c6ca5a1f3396bd2e3fad5bfa","subject":"Merge \"Changed /tmp/migrate-db.rc to /root/migrate-db.rc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ce3dd3839ba4df38c6ca5a1f3396bd2e3fad5bfa"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2023-10-18 15:30:26.000000000","tz":0},"committer":{"name":"Jan Gutter","email":"jan.gutter@workday.com","date":"2024-10-07 11:00:55.000000000","tz":0},"subject":"DNM -- testing fix in external PR","message":"DNM -- testing fix in external PR\n\nThis is testing the proposed fix (code only) in\nhttps://github.com/sapcc/placement/pull/5/files to see which\ntests (if any) are tripped.\n\n( this is a slightly modified version of\nhttps://review.opendev.org/c/openstack/placement/+/898808 )\n\n- adding in the test fixture update from:\n  https://github.com/sapcc/placement/pull/5/commits/61164db5e07b7bc4146da63dc3051220752ccb52\n\nChange-Id: I41392ad959fa5c8bbcdeb5f0b7ea9819dea7690b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/4ecb80429f802ecbbeb0ca25f061597bc4d63b9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/4ecb80429f802ecbbeb0ca25f061597bc4d63b9c"}]},"branch":"refs/heads/master"},"9b9007afa395241e4eca80b9728c85b108f89a4e":{"kind":"REWORK","_number":5,"created":"2024-11-08 18:58:49.000000000","uploader":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"ref":"refs/changes/43/926743/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/43/926743/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/43/926743/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/43/926743/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/43/926743/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/43/926743/5"}}},"commit":{"parents":[{"commit":"f8587f2f52fb95f790c91a2526cabcb9b5c065e6","subject":"Merge \"Remove Python 3.8 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f8587f2f52fb95f790c91a2526cabcb9b5c065e6"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2023-10-18 15:30:26.000000000","tz":0},"committer":{"name":"Jan Gutter","email":"jan.gutter@workday.com","date":"2024-11-08 18:58:19.000000000","tz":0},"subject":"DNM -- testing fix in external PR","message":"DNM -- testing fix in external PR\n\nThis is testing the proposed fix (code only) in\nhttps://github.com/sapcc/placement/pull/5/files to see which\ntests (if any) are tripped.\n\n( this is a slightly modified version of\nhttps://review.opendev.org/c/openstack/placement/+/898808 )\n\n- adding in the test fixture update from:\n  https://github.com/sapcc/placement/pull/5/commits/61164db5e07b7bc4146da63dc3051220752ccb52\n\nChange-Id: I41392ad959fa5c8bbcdeb5f0b7ea9819dea7690b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9b9007afa395241e4eca80b9728c85b108f89a4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9b9007afa395241e4eca80b9728c85b108f89a4e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
