)]}'
{"id":"openstack%2Fplacement~926935","triplet_id":"openstack%2Fplacement~master~Iaf0c3eeb75955c56c69616fc83733659f41637f4","project":"openstack/placement","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-09-03 21:34:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaf0c3eeb75955c56c69616fc83733659f41637f4","subject":"Bump oslo.policy version to enable new RBAC by default","status":"MERGED","created":"2024-08-22 16:53:43.000000000","updated":"2024-09-03 21:36:57.000000000","submitted":"2024-09-03 21:34:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926935-secure-rbac","meta_rev_id":"53d97692a97bd107effa48fbc8001d19027d7235","_number":926935,"virtual_id_number":926935,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-03 21:34:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2024-09-03 19:27:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2024-08-22 18:34:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2024-09-03 19:27:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-22 18:17:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-22 18:34:36.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2024-09-03 19:27:18.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"0547f3ebf203e41957191a105fadf390100f6e37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-22 16:53:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a85a2102c1783bd53302a0d8ded6aff1b9750f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 18:17:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e928be121e4a4635bb4a85d43ca2287c\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3927ba1d76824e92b7b4ca84ca79cb0b : SUCCESS in 3m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/84e744734f4d4aae820780a98808c9dc : SUCCESS in 1h 06m 21s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/afcdc731b2cb4a968700646817711454 : SUCCESS in 1h 18m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d4ad93fc906c4f669a54ee554ed4e0d4 : SUCCESS in 1h 08m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e336256e6de240d8bd6a57df392c51ef : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48d6f12d37904e09865ce660e6e6b4ae : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/445ee9cc68c44644be7a2384a738e020 : SUCCESS in 4m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/636c4bc361e94ac690fc65f88c0d8647 : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e0cb915a3897464db244f444501fc859 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee01c8c189814a6cac6ae34f6601dbb7 : SUCCESS in 7m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0ec01e8b5df54e0388f00997fca75fc3 : SUCCESS in 7m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dbae4c22467246889c024047a80c4a01 : SUCCESS in 5m 33s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1f19306451c145b3a797e0bcf52bfcd5 : SUCCESS in 20m 02s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/b9439c893dc44cb3803fa1ecc3a56e71 : SUCCESS in 22m 05s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/7d6cfa311eeb451386a152e16f3ead72 : FAILURE in 3m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ff0a288d799404e9cf8109ab1a6f22c : SUCCESS in 49m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b55d20d21e39ce5d9160e64cf1df5674de25beee","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-08-22 18:34:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f2220cf31163f128fea3171ba5b3011ba3a87a50","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-03 19:06:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"253e735089724218222879d434f55a9d137e30bc","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-09-03 19:27:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"80acb24dbee402dd799aa7e3e418e39e5c35af50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 19:27:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b722ed05a4912b536cac1bf5fd98f51f4bd3353a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 21:34:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f729970f73b44b7814377cf531df30a\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/dd2330b380374a40a1a34af9bbf36c35 : SUCCESS in 3m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/c4541ed840f247258a691d580114d743 : SUCCESS in 1h 21m 59s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/1afd07d84ebd42d9bc4397d8388194bc : SUCCESS in 1h 32m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7af45309a3e54d91ab76b3a9fcdfd2f8 : SUCCESS in 58m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b272e81ae51a490c8db6108fe7cb1e91 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a56da3cf1a0d408e995ac6a45d44eb16 : SUCCESS in 5m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dcde196ec18645a98b12c72ffba777b9 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baf4f1e66a3341009166159f3f12f804 : SUCCESS in 7m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c4ab005a12264a498e624b6292c74a03 : SUCCESS in 5m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b836a908e14f4344bd3fbda0a3c8c1d8 : SUCCESS in 5m 15s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1350c11beacd48398dee8b90de3ed279 : SUCCESS in 21m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/57e840799da041f390f3a3b80867fc17 : SUCCESS in 55m 19s","accounts_in_message":[],"_revision_number":1},{"id":"313c76f82eca543a5dd6e42e60fa6380c9926b16","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 21:34:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"53d97692a97bd107effa48fbc8001d19027d7235","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 21:36:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a450c0b0aa341858d59334e7d57e55e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddd3d63738764b7eab3f4d0201a29262 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5a340f75481d9aa7a0eff33add75b42d9a577ecf","revisions":{"5a340f75481d9aa7a0eff33add75b42d9a577ecf":{"kind":"REWORK","_number":1,"created":"2024-08-22 16:53:43.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/35/926935/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/35/926935/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/35/926935/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/35/926935/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/35/926935/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/35/926935/1"}}},"commit":{"parents":[{"commit":"95631d77290a05a638aa8b5c1d83bac1766c55f4","subject":"Update test to use service role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/95631d77290a05a638aa8b5c1d83bac1766c55f4"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-22 16:51:24.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-22 16:51:24.000000000","tz":-420},"subject":"Bump oslo.policy version to enable new RBAC by default","message":"Bump oslo.policy version to enable new RBAC by default\n\noslo.policy 4.4.0 enable the RBAC new defaults by default\nwhich enable them for placement also.\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/925464\nChange-Id: Iaf0c3eeb75955c56c69616fc83733659f41637f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/5a340f75481d9aa7a0eff33add75b42d9a577ecf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/5a340f75481d9aa7a0eff33add75b42d9a577ecf"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"95631d77290a05a638aa8b5c1d83bac1766c55f4","is_merged_in_target_branch":true,"change_id":"I87922b4045fe0a2cab834f04d5d98a969821102b","change_number":926870,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
