)]}'
{"id":"openstack%2Fplacement~933116","triplet_id":"openstack%2Fplacement~master~Ia140968dd184daf25759336838cabf45b15d0156","project":"openstack/placement","branch":"master","topic":"audit-middleware","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-16 17:40:10.000000000","reason":"Change was submitted"},"9542":{"account":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"last_update":"2026-09-16 17:40:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia140968dd184daf25759336838cabf45b15d0156","subject":"CADF audit support for Placement API","status":"MERGED","created":"2024-10-23 14:29:17.000000000","updated":"2026-09-16 17:41:43.000000000","submitted":"2026-09-16 17:40:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933116-audit-middleware","meta_rev_id":"87b3ab0bb528e9ff51a391e50014fb9270c3639b","_number":933116,"virtual_id_number":933116,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-16 17:40:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2026-09-16 15:20:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2026-09-16 13:58:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2026-09-16 15:20:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":35392,"name":"Oleksandr Klymenko","display_name":"Oleksandr Klymenko","email":"oklymenko@mirantis.com","username":"oklymenko"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-23 14:29:17.000000000","updated_by":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"reviewer":{"_account_id":35392,"name":"Oleksandr Klymenko","display_name":"Oleksandr Klymenko","email":"oklymenko@mirantis.com","username":"oklymenko"},"state":"CC"},{"updated":"2024-10-23 15:46:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-15 21:22:47.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-09-16 15:20:09.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"489c1036109bbf44b6268a67b7ce9707b3b456d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2024-10-23 14:29:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea21fc972d77a7ab41d0581d66ceed723a557575","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 15:46:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b911514431a54fb2b5cc1c4ce6cdb522\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/51cd29ec81e44a5cb9d0b2e0b3d4ab27 : SUCCESS in 1m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/fb9283163ceb4a7bbae1729942a29d82 : SUCCESS in 1h 10m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2373a81189e343efaafe3981158ffb39 : SUCCESS in 1h 08m 59s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/7e33a1ecedba4233af4e028ae4f68fc3 : SUCCESS in 1h 05m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f2970a4d44b4f20a7ec8d7909703302 : SUCCESS in 44m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23d0d842961c4553a263333c16ef56f6 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b08b1e27f2e54013b6a577e9aca3be07 : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0da5e572522c44349fa1ed1018fce84e : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/67a6241110b449b4af42bd78f10d6002 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f336afabbca8499b84de7e91fa5a940f : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b182723aac1345af8edf8f24ae3a33ac : SUCCESS in 3m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e750af731afd4ea1a9e761dabe15f0a0 : SUCCESS in 5m 54s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b4d2685149ec483a920e2e6b721a18b1 : SUCCESS in 8m 35s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2441a81cbc6a40019443fbf3880c717b : SUCCESS in 25m 03s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/41c606b022c34ed48ca4c111023bddbf : SUCCESS in 15m 29s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/b71f04ec57fe45dd8749c53510b1dc9c : FAILURE in 3m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83f7ce9d0a6b4000a3b5888e0ed5fa19 : SUCCESS in 36m 06s","accounts_in_message":[],"_revision_number":1},{"id":"62eefb9f042809f7d29a2b0fd9e29060c746e768","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-15 21:22:19.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"c131a3e188f9f71814c84095e9ba2473e723d0fa","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-15 21:22:47.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4508c65c368cd3bd20f06c7d51f6a79720542dc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 22:51:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a582167959c64944b5f9d2b37106a233\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b7420bed3629489398b3ebb59b32016a : SUCCESS in 2m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f8acdbace64f4e8c8fc45bab2111d94b : FAILURE in 3m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/1a26f65cea9e44ad9d5a1d57e39be6ef : SUCCESS in 1h 14m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0318bae927e84b5aa271def85bc1e86b : SUCCESS in 1h 01m 57s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/cd5c6815b9fc4bafad23798fe19f1289 : SUCCESS in 1h 23m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f61a138b963a418086f66719ffacdf17 : SUCCESS in 1h 01m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b9288b5787744dfa74e6fd0e3c88987 : SUCCESS in 15m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba33023614104d6ba964411b1e625a50 : SUCCESS in 5m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/68d60cf7f6d94ea5b7366c17ef1da790 : SUCCESS in 3m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ca792593ec5945cfb53a12527fc695bd : SUCCESS in 4m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/af3190d6c0ff4763923a61b825aa5352 : FAILURE in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8abeed2cac54020bfaf8753b9df28ee : SUCCESS in 10m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad8c0ce056de4502833023cc55a451d1 : SUCCESS in 10m 40s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/700bb67eed9d4bc69d88c6f7f25eeffd : SUCCESS in 4m 04s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3c0640a4b75b45a8b23d1171e260d0c9 : SUCCESS in 6m 20s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/ca248bb1d64247cbbbd3fe7db5473e8a : SUCCESS in 8m 39s (non-voting)\n- placement-nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e06a612717cd4d4698d0abdd41b82d5d : SUCCESS in 22m 39s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/80cf8fe79fad449bac1f758c7ea316e0 : FAILURE in 4m 01s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/a56ee8272cd441e1bb1efed9723b1985 : FAILURE in 3m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/277ea6af7bdb471d9fb01a44833ffe02 : SUCCESS in 58m 03s","accounts_in_message":[],"_revision_number":2},{"id":"a7f4852c60b3eabc380c2711a4b2bd507a08027b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-16 13:55:27.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"a6f012c3c3fe228abbf1fed4e0deed51a6d58c54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-16 13:56:28.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"bc8dc52abb58fa28029124e6eddf1edfc4d92205","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-16 13:58:53.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"abc3aaaffa0bf2e36f0eeb9f517d69e1a224b47f","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-09-16 15:20:09.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a325798992cce328115d36ccbe39dab17be10fb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 15:46:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fd3b6ab344744bf9cd20b348e107739\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9e2f9c40bb264c03a9ce3e8e7030eea8 : SUCCESS in 2m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/eef9e240bdcc49498ae5c053add755a3 : SUCCESS in 4m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/801c9c26aa5f43999fcbb7b0b38da0ac : SUCCESS in 1h 02m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eb5d2e3b94474c68824a0d71648fbef9 : SUCCESS in 1h 13m 53s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/7bb6e1306c714c0a8a68ed6c8b20a9b8 : SUCCESS in 1h 35m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a80a91a9d5764ea3911c0888937aa6f8 : SUCCESS in 55m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5354097881534f2da774fd0db73c3d96 : SUCCESS in 16m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60544977215c4c0da88f6ff5c2a1cb77 : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2576a330f9de49f7a259a0742360c989 : SUCCESS in 5m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3df64a2edf62411d86f530405a588787 : SUCCESS in 3m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/29735ff20f6541aca7305a5c3a7dd453 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4763cd09f54843d28e8c42aa52d4aea1 : SUCCESS in 8m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ca34a81437d4423856ec353d72a9010 : SUCCESS in 4m 49s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9d2a4c3cf7254fc491ac960407a8111a : SUCCESS in 7m 03s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ff3d22a9ab284b95953bd37a1d582092 : SUCCESS in 4m 12s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/74137b4965cc47198da3837347b94baf : SUCCESS in 8m 40s (non-voting)\n- placement-nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/19dde62cb20e47ab8df52d6ecbe0ed37 : SUCCESS in 23m 10s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/a0bcfb1e756d42dcaa5368a42c8b6a44 : FAILURE in 3m 10s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/1a071c2f0f8d459490bc9f60cae4edca : FAILURE in 3m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9bd65536aa9e4ce781b92de531a473f8 : SUCCESS in 1h 02m 26s","accounts_in_message":[],"_revision_number":4},{"id":"40588321194d51b7c36e7aaae63c5f049dc5ad44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 15:47:25.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e97f94e52b43aff13ed6d19f1703efdd63f0951e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 17:40:09.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21a5624542be4d5381df56a080888282\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0c780df67df94ae4bf3cc7333761c424 : SUCCESS in 4m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0d9c0df2c69544b1888c941c374efbcc : SUCCESS in 2m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/4ec70bd50cc04676b531c568f8fc406a : SUCCESS in 1h 05m 19s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/f034380e07404215b062ced3c46f7664 : SUCCESS in 1h 46m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0ebaca955ccd498db366a8ef76515c93 : SUCCESS in 1h 05m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff02b4ca5c9840e6be855823a9ab73cf : SUCCESS in 44m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/603c5cbcd588482a84c88ee042a370a3 : SUCCESS in 6m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/78268f2f4dff46b39d82d252bad93be4 : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e3caa5ee31ee4e4b99690205062ca411 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/153a7b4f629c46cf972300302908bfc4 : SUCCESS in 9m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39078f64dc034d368561a53630696f65 : SUCCESS in 10m 13s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e8aa84d9cbca4b7dbbc9f4a19c3f75a8 : SUCCESS in 5m 57s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b8addab0462248d0be519440689427e5 : SUCCESS in 6m 13s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/979b183e57e24e59b15ab0136a58f95c : SUCCESS in 9m 31s (non-voting)\n- placement-nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/68cfbe5f527349dfb4e9777b6bce4f2e : SUCCESS in 21m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/433edaac8bff41dd911ec51a924d7d7c : SUCCESS in 57m 54s","accounts_in_message":[],"_revision_number":4},{"id":"9a650820bb49bddc5a80ba1e12aad995f6880847","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 17:40:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"87b3ab0bb528e9ff51a391e50014fb9270c3639b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 17:41:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/134f83345d374fa39abbe86cf3a1bfdb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f823f4e163443cea42f9d4794bcd4ba : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06576110315c4b1ea2f2a239c507af5b : SUCCESS in 49s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7ec4d1c49753e4a897250595701fecb39ce087f9","revisions":{"d0c0dcf9ede0379312d7c511f05195a0bdaf50fa":{"kind":"REWORK","_number":1,"created":"2024-10-23 14:29:17.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/16/933116/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/16/933116/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/16/933116/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/16/933116/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/16/933116/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/16/933116/1"}}},"commit":{"parents":[{"commit":"f8587f2f52fb95f790c91a2526cabcb9b5c065e6","subject":"Merge \"Remove Python 3.8 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f8587f2f52fb95f790c91a2526cabcb9b5c065e6"}]}],"author":{"name":"oklymenko","email":"oklymenko@mirantis.com","date":"2023-04-12 11:29:56.000000000","tz":180},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2024-10-23 14:29:10.000000000","tz":0},"subject":"CADF audit support for Placement API","message":"CADF audit support for Placement API\n\nthis patch adds support for audit middleware from\nkeystone middleware to Placement.\n\nWhen enabled and given correct audit map file,\nPlacement service will emit CADF audit notifications.\n\nChange-Id: Ia140968dd184daf25759336838cabf45b15d0156\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d0c0dcf9ede0379312d7c511f05195a0bdaf50fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d0c0dcf9ede0379312d7c511f05195a0bdaf50fa"}]},"branch":"refs/heads/master"},"2c7b038f06fe65a00fe3f3ec340fedcb55365638":{"kind":"REWORK","_number":2,"created":"2026-09-15 21:22:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/16/933116/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/16/933116/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/16/933116/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/16/933116/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/16/933116/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/16/933116/2"}}},"commit":{"parents":[{"commit":"8e91e4d7aad1c6a642f80ec6387ce08d86f7c26d","subject":"Update master for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/8e91e4d7aad1c6a642f80ec6387ce08d86f7c26d"}]}],"author":{"name":"oklymenko","email":"oklymenko@mirantis.com","date":"2023-04-12 11:29:56.000000000","tz":180},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-15 21:22:09.000000000","tz":60},"subject":"CADF audit support for Placement API","message":"CADF audit support for Placement API\n\nthis patch adds support for audit middleware from\nkeystone middleware to Placement.\n\nWhen enabled and given correct audit map file,\nPlacement service will emit CADF audit notifications.\n\nChange-Id: Ia140968dd184daf25759336838cabf45b15d0156\nSigned-off-by: oklymenko \u003coklymenko@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/2c7b038f06fe65a00fe3f3ec340fedcb55365638"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/2c7b038f06fe65a00fe3f3ec340fedcb55365638"}]},"branch":"refs/heads/master"},"bc9adb82cdd89677efd383fa773f2dd4b9e4ac1f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-09-16 13:55:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/16/933116/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/16/933116/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/16/933116/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/16/933116/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/16/933116/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/16/933116/3"}}},"commit":{"parents":[{"commit":"3d3724ea24bf3e1950e638d5419460245d621523","subject":"Fix tests on Python 3.14","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/3d3724ea24bf3e1950e638d5419460245d621523"}]}],"author":{"name":"oklymenko","email":"oklymenko@mirantis.com","date":"2023-04-12 11:29:56.000000000","tz":180},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-16 13:55:27.000000000","tz":0},"subject":"CADF audit support for Placement API","message":"CADF audit support for Placement API\n\nthis patch adds support for audit middleware from\nkeystone middleware to Placement.\n\nWhen enabled and given correct audit map file,\nPlacement service will emit CADF audit notifications.\n\nChange-Id: Ia140968dd184daf25759336838cabf45b15d0156\nSigned-off-by: oklymenko \u003coklymenko@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/bc9adb82cdd89677efd383fa773f2dd4b9e4ac1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/bc9adb82cdd89677efd383fa773f2dd4b9e4ac1f"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"8e91e4d7aad1c6a642f80ec6387ce08d86f7c26d","ours":"2c7b038f06fe65a00fe3f3ec340fedcb55365638","theirs":"3d3724ea24bf3e1950e638d5419460245d621523","merge_strategy":"recursive","contains_conflicts":false}},"7ec4d1c49753e4a897250595701fecb39ce087f9":{"kind":"REWORK","_number":4,"created":"2026-09-16 13:56:28.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/16/933116/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/16/933116/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/16/933116/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/16/933116/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/16/933116/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/16/933116/4"}}},"commit":{"parents":[{"commit":"3d3724ea24bf3e1950e638d5419460245d621523","subject":"Fix tests on Python 3.14","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/3d3724ea24bf3e1950e638d5419460245d621523"}]}],"author":{"name":"oklymenko","email":"oklymenko@mirantis.com","date":"2023-04-12 11:29:56.000000000","tz":180},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-16 13:56:28.000000000","tz":0},"subject":"CADF audit support for Placement API","message":"CADF audit support for Placement API\n\nthis patch adds support for audit middleware from\nkeystone middleware to Placement.\n\nWhen enabled and given correct audit map file,\nPlacement service will emit CADF audit notifications.\n\nChange-Id: Ia140968dd184daf25759336838cabf45b15d0156\nSigned-off-by: oklymenko \u003coklymenko@mirantis.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/7ec4d1c49753e4a897250595701fecb39ce087f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/7ec4d1c49753e4a897250595701fecb39ce087f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
