)]}'
{"id":"openstack%2Fplacement~937273","triplet_id":"openstack%2Fplacement~master~I4f8c8491d21426310a82df7186b7689b2dad253e","project":"openstack/placement","branch":"master","topic":"bug/2070257","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2024-12-12 03:26:32.000000000","reason":"\u003cGERRIT_ACCOUNT_4690\u003e replied on the change","reason_account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2024-12-11 13:07:22.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2025-01-08 21:44:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4f8c8491d21426310a82df7186b7689b2dad253e","subject":"Factor out allocation candidate generation strategy","status":"MERGED","created":"2024-12-06 17:39:55.000000000","updated":"2025-01-08 21:45:29.000000000","submitted":"2025-01-08 21:44:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"937273-bug/2070257","meta_rev_id":"a029bea5ae81b2bfc0a73a4f1b23643a8a4dfdd3","_number":937273,"virtual_id_number":937273,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-08 21:44:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":2,"date":"2024-12-12 03:26:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":2,"date":"2025-01-08 19:24:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":1,"date":"2025-01-08 19:24:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-06 17:41:54.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2024-12-06 19:58:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-09 15:50:23.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"},{"updated":"2025-01-08 19:24:12.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"0e41eee47f205973379129bd2a123e810c8f4fc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-06 17:39:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36e0c9d383ca2c868080e59da51fbf6ac9ed7e4e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-06 17:49:39.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"179e5890484e2bdf2787c9c102348cac60658e1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 19:58:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e0c87b2c44b46569d7b4da80605aa1d\n\n- grenade https://zuul.opendev.org/t/openstack/build/cbfad318024e421aaff88f1082ca81ff : SUCCESS in 48m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b614e06146794e3e8960e1f3f78a3c27 : SUCCESS in 1h 03m 35s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/0bf2dc78f37e4fa99bdb353f5008cd99 : SUCCESS in 1h 42m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9274bdf5b96141618f845a04272b5976 : SUCCESS in 1h 15m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff119f1f56a94ee9b5b6c50d2e526671 : SUCCESS in 16m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77429c8e295d463b9902f2ddfb96f6e6 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d2359f65b664d3bab1c58421e77bd0d : SUCCESS in 5m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/259e601610134e3aa632455ec1d5809b : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/716c7d4d7b674c1a9447bc75290b392e : SUCCESS in 7m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b89673df653c4834b8d61c239343727b : SUCCESS in 7m 40s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9966aa919a6c494b8a0c42b15cc3aa9e : SUCCESS in 10m 33s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c3cbc9233f684b7caa32c7782e306c70 : SUCCESS in 23m 07s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/cf41ae9becce4f0491176b027a5f7b63 : SUCCESS in 19m 35s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/2e3e27b17cc2439bb6040105ad705a3e : FAILURE in 3m 15s (non-voting)\n- tempest-integrated-placement-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7acbc315a49245a09e5dbede2bdf7ac4 : SUCCESS in 1h 41m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11f5c48679f140eeb5e2bc1ebd1501ba : SUCCESS in 1h 00m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d428e0d2d168191b47464e42893085a0ecf6ce0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-09 13:40:38.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1c702d9dab7f3372058b36fad8d9d6e533320d39","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-12-09 15:50:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f0ee2f0019404164e627393615580112a7ca9b1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 16:25:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca7f4e4d1dbf47249958c2ce8ed02dda\n\n- grenade https://zuul.opendev.org/t/openstack/build/b4a1947b517c4facabee9f4d8cb9c101 : SUCCESS in 43m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6676426f76444aa7bb99596e65948ea5 : SUCCESS in 58m 28s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/7daaca3133144cd392f74bc1f5d0b3e6 : SUCCESS in 1h 42m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f181577459243b89b76117e283c4abe : SUCCESS in 1h 15m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1280498de05d413b81905bc201adf115 : SUCCESS in 14m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5790d1ebf226431fbbb85e938b1b1c73 : FAILURE in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c21dc52a49c6434eaacbdbdea78cc777 : SUCCESS in 5m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/259dab50d19041df87fb34b125fda332 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80a638e2b5634b6ca3f62afa6afae13c : SUCCESS in 7m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f899e435eb0a4558b165acc8a5fc198b : SUCCESS in 6m 29s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a88e9bd026d94f969345ae2eba88932f : SUCCESS in 9m 19s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d9fc5f47e6404496ac9099fe9f56a9ad : SUCCESS in 24m 05s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/b80a43e40040470997839c6c564aa6f4 : SUCCESS in 17m 09s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/9f9c467701634b2bb95f70983148ffb9 : FAILURE in 3m 04s (non-voting)\n- tempest-integrated-placement-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b022c63141df4ff1b6c306b50dc02be1 : SUCCESS in 1h 34m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f98618997894cd6afe33ccc8ab5bd33 : SUCCESS in 36m 33s\n\nWarning:\n  Comments left for invalid file placement/tests/functional/test_allocation_candidates.py","accounts_in_message":[],"_revision_number":2},{"id":"d2a9a2e08cb4ae39bac195ae07fd8771e784bdef","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-12-09 23:53:53.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ff694447a956f79cb0df3ac63fa77df8b6398c3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-11 12:38:19.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"81b0460cbe7b57ff79b0696a82b18736eef9aee1","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-11 12:38:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b6c9f419803979a02eafc01a2d72ca5b681aa70a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-12-11 13:07:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b4a3ee8858e0bbe095fea895bc9b00515d05555e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 14:16:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3c5fba6e53945d5931fa0937acc29bf\n\n- grenade https://zuul.opendev.org/t/openstack/build/2627a16f7b614237ae3580cf800af588 : SUCCESS in 59m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b2b6d3be0e154ea88fdfc64cfabe2410 : SUCCESS in 1h 05m 35s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/e0aa6fe1deef4791968dbbdc97bc93f4 : SUCCESS in 1h 37m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a25ed7151935416bbfee4e21e554275b : SUCCESS in 1h 06m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c556c97e1724fc2aba3ea6c43fbedd7 : SUCCESS in 14m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b899dbb2fea46d59ec4aaaff037db78 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05c685f272c643a78fb0f89bcdf84651 : SUCCESS in 4m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/89bba4d457be41739ae550d51480483e : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/833d80a7b8f7464b8fe8c9f83634584d : SUCCESS in 7m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c784226806b3456bb7544ed95859fedf : SUCCESS in 6m 22s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/dbe703ed719b46a1bcde376efacf19a8 : SUCCESS in 9m 16s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/21a3f36d120b42bb908b9351c66bcf40 : SUCCESS in 24m 49s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/fad5d52ee4d34c0b86313777b828f564 : SUCCESS in 22m 03s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/bb55370dad384f88a8bfa0ec6177d204 : FAILURE in 2m 21s (non-voting)\n- tempest-integrated-placement-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50c7f0f8dd0e48b9bcb757fb2fb0336a : FAILURE in 1h 26m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/885354ce216447fa993ab4e710fb6751 : SUCCESS in 48m 43s","accounts_in_message":[],"_revision_number":3},{"id":"d7bd94c95e58bde618a3aafe1fa2bf25d7ab6c5f","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-12-12 03:26:32.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"71ddba3afe3aafbd42b183549b65a025a87a6ac0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 04:51:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ce914259ae64ab3ad2745573f25e94e\n\n- grenade https://zuul.opendev.org/t/openstack/build/fa2f7d6780d9420f97ce71ae9e6e2d30 : SUCCESS in 49m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/61e8c8752c4a42609f327847568183e0 : SUCCESS in 52m 32s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/bd8b1d72d7e34608b6f0caddce1e7ba0 : SUCCESS in 1h 24m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c4ab41b69f54abb883bae9a9e9001f5 : SUCCESS in 53m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef166abfd8bc4c6182f484a39e1982cd : SUCCESS in 9m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a88840008f99411c9c3ef19af160f2a9 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f743102021404929b2a2385a51076073 : SUCCESS in 5m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0a62c80ea3034cec8834d1df491ab88b : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2884783334314495a4bb10a91c04a0ba : SUCCESS in 7m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c1ea8ce31a5b41cb8570e194741be290 : SUCCESS in 6m 55s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b488b8f27ef24aa0ba8e655131340b5a : SUCCESS in 5m 54s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c74f5395156f4ed7ad88deff91d80a35 : SUCCESS in 20m 55s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/6b8579628a9a45c4ba512f5c26d1372d : SUCCESS in 10m 14s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/663fd834b5e54df6b6773ffe80017786 : FAILURE in 3m 07s (non-voting)\n- tempest-integrated-placement-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/661ba6d0fb154fd380505799725519b5 : SUCCESS in 55m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d1a44b2dc6f405aa5f4b5b0a5fea01f : SUCCESS in 1h 00m 46s","accounts_in_message":[],"_revision_number":3},{"id":"1d63c4b8a5050290629ca91b55f47e9334d9eaf0","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-07 20:05:55.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"898c4f8f130db61c9ffad2fecea4f5126fcc7c53","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-08 19:24:12.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1204aed45a8fe9fa3626dc39badfeca3c3eed091","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 19:57:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d79043ef11ad9577529871f7651a114e3130a298","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 21:44:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a505b68c76254beca27fb23ebdf0a913\n\n- grenade https://zuul.opendev.org/t/openstack/build/6d2619e6b951472f858ae9307f9d4e3a : SUCCESS in 57m 26s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/1785afe94cbd4f94aff1540f66439791 : SUCCESS in 1h 44m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4b5950e62ca24067b5ef26e51898ce53 : SUCCESS in 57m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18bb594bdfc14c91a0f3cdd6aab942f0 : SUCCESS in 1h 04m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e324b7dd0c58442ebd3a88c1838745a9 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53353b855a2546e3b48bba6beaa0c97a : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/943c1457148446b38ff182b9ed344bc0 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a45c6e8023514c60aa9faa15b248c10e : SUCCESS in 7m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3f97cb3e54284bf6892f2f571a449678 : SUCCESS in 4m 09s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a262769d885047a589f4e400755d5627 : SUCCESS in 7m 28s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8804c1f2c9aa485787630526a59e0787 : SUCCESS in 21m 30s\n- tempest-integrated-placement-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/13c528fc19df48f896dc3cc7e9649a20 : SUCCESS in 1h 08m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5792740858ca4664873e5f81b14cea3c : SUCCESS in 1h 04m 09s","accounts_in_message":[],"_revision_number":3},{"id":"a3819721163b3fa22b430d4e08adc422d88b64c0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 21:44:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a029bea5ae81b2bfc0a73a4f1b23643a8a4dfdd3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 21:45:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bce5126647e04a08817f4ce83ab199dc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9614e8aa6e4b4ae0b1f47aaf16c842c7 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"04718d65b60e817b5f47c5a3f50e302fe2302dfd","revisions":{"c52310a9a563b2a5e2bd720ebe7ff2c44e55afea":{"kind":"REWORK","_number":1,"created":"2024-12-06 17:39:55.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/73/937273/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/73/937273/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/73/937273/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/73/937273/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/73/937273/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/73/937273/1"}}},"commit":{"parents":[{"commit":"8589498abb9090713185af6dcdc8c84ad29ef774","subject":"Add a global limit on the number of allocation candidates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/8589498abb9090713185af6dcdc8c84ad29ef774"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-06 15:47:53.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-06 16:15:27.000000000","tz":60},"subject":"Factor out allocation candidate generation strategy","message":"Factor out allocation candidate generation strategy\n\nThis is a pure refactor to pull out the logic of assembling an\nallocation candidate.\n\nThe _get_areq_list_generators returns a set of\ngenerators, one per root (anchoring) provider, that can generate all the\nviable candidates for that root by applying a product on the solutions\nfor each request groups.\n\nThe _generate_areq_lists implements the strategy of consuming these\ngenerators. The current behavior is that placement consumes the\ngenerator for the first root fully before moving to the next\n(depth-first).\n\nA later patch will change _generate_areq_lists to be able to apply\ndifferent strategies, like round-robin easily.\n\nRelated-Bug: #2070257\nChange-Id: I4f8c8491d21426310a82df7186b7689b2dad253e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/c52310a9a563b2a5e2bd720ebe7ff2c44e55afea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/c52310a9a563b2a5e2bd720ebe7ff2c44e55afea"}]},"branch":"refs/heads/master"},"96a40311c9a5a1e1a6d1e6a3f331fb242a3c1a57":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-12-09 13:40:38.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/73/937273/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/73/937273/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/73/937273/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/73/937273/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/73/937273/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/73/937273/2"}}},"commit":{"parents":[{"commit":"e0a35f2b9ae1f0349a4fb9d9f8d78144f74db9ee","subject":"Add a global limit on the number of allocation candidates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e0a35f2b9ae1f0349a4fb9d9f8d78144f74db9ee"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-06 15:47:53.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-09 13:00:56.000000000","tz":60},"subject":"Factor out allocation candidate generation strategy","message":"Factor out allocation candidate generation strategy\n\nThis is a pure refactor to pull out the logic of assembling an\nallocation candidate.\n\nThe _get_areq_list_generators returns a set of\ngenerators, one per root (anchoring) provider, that can generate all the\nviable candidates for that root by applying a product on the solutions\nfor each request groups.\n\nThe _generate_areq_lists implements the strategy of consuming these\ngenerators. The current behavior is that placement consumes the\ngenerator for the first root fully before moving to the next\n(depth-first).\n\nA later patch will change _generate_areq_lists to be able to apply\ndifferent strategies, like round-robin easily.\n\nRelated-Bug: #2070257\nChange-Id: I4f8c8491d21426310a82df7186b7689b2dad253e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/96a40311c9a5a1e1a6d1e6a3f331fb242a3c1a57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/96a40311c9a5a1e1a6d1e6a3f331fb242a3c1a57"}]},"branch":"refs/heads/master"},"04718d65b60e817b5f47c5a3f50e302fe2302dfd":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-12-11 12:38:19.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/73/937273/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/73/937273/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/73/937273/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/73/937273/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/73/937273/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/73/937273/3"}}},"commit":{"parents":[{"commit":"9351241f91f80cb7dc45c4c0daab929b5dfc5c4b","subject":"Add a global limit on the number of allocation candidates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9351241f91f80cb7dc45c4c0daab929b5dfc5c4b"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-06 15:47:53.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-11 12:27:35.000000000","tz":60},"subject":"Factor out allocation candidate generation strategy","message":"Factor out allocation candidate generation strategy\n\nThis is a pure refactor to pull out the logic of assembling an\nallocation candidate.\n\nThe _get_areq_list_generators returns a set of\ngenerators, one per root (anchoring) provider, that can generate all the\nviable candidates for that root by applying a product on the solutions\nfor each request groups.\n\nThe _generate_areq_lists implements the strategy of consuming these\ngenerators. The current behavior is that placement consumes the\ngenerator for the first root fully before moving to the next\n(depth-first).\n\nA later patch will change _generate_areq_lists to be able to apply\ndifferent strategies, like round-robin easily.\n\nRelated-Bug: #2070257\nChange-Id: I4f8c8491d21426310a82df7186b7689b2dad253e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/04718d65b60e817b5f47c5a3f50e302fe2302dfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/04718d65b60e817b5f47c5a3f50e302fe2302dfd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
