)]}'
{"id":"openstack%2Fplacement~937274","triplet_id":"openstack%2Fplacement~master~I3218bbbecb6b2dd6e94fba270cf3084f278b404f","project":"openstack/placement","branch":"master","topic":"bug/2070257","attention_set":{},"removed_from_attention_set":{"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2025-01-13 12:34:02.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3218bbbecb6b2dd6e94fba270cf3084f278b404f","subject":"DNM: test with breadth-first in tempest","status":"ABANDONED","created":"2024-12-06 17:39:55.000000000","updated":"2025-01-13 12:34:02.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"069139eba6b7c945a4012fcd40619c2cf3d9cae0","_number":937274,"virtual_id_number":937274,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-01-09 10:13:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-06 19:30:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d5746f13ae284855bbeb8a4f1491fcca359aac06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-06 17:39:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ac444e8a5eeb04603367fda2709351560f15bb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 19:30:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64ac590474ab430dad9e79e51d5b310b\n\n- grenade https://zuul.opendev.org/t/openstack/build/fccfd7d0f3f84ac88e580ac373807289 : SUCCESS in 1h 06m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c0f321c2b5cb4bb8bd8974fa3e7ac96b : SUCCESS in 56m 48s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/f84fa68c01474453a9bde433084049c9 : SUCCESS in 1h 31m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45c164bc945b48c4bbf6655b5aa2ac9e : SUCCESS in 1h 02m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c6805050e964291828f97f9c67ade03 : SUCCESS in 13m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad50543988344d4c8d91b47dd7ffe780 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6231c316ee648128a581e89bf5437aa : SUCCESS in 8m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ab1bde27aaa444a38026ef4e0de91a7a : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6a7c26e3f564790aabb2913eae0f138 : SUCCESS in 6m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dd01a6490fdf451d8e635039559aa6f1 : SUCCESS in 6m 20s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/99b8eff4a1c44606929567c47b3996ab : SUCCESS in 10m 43s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/332551f67da84c1c85c28b6bb11dcdea : SUCCESS in 23m 38s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/1a767401ad714b98b6e3e2952a22e91e : SUCCESS in 24m 01s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/489510fde0aa461d9a3e9c54cc004a45 : FAILURE in 3m 16s (non-voting)\n- tempest-integrated-placement-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/833ea6a7b1e141d3b8f773214e58fada : SUCCESS in 1h 45m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/19b20227074042ef910c453c124c7d70 : SUCCESS in 55m 58s","accounts_in_message":[],"_revision_number":1},{"id":"56fc5c42c9684ee6a8ca112e266e834ff3e97260","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-09 12:20:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f0234375974eba98b266b29faedad311d738341e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-09 13:40:38.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"80ff52cfd2ad81066c4ceeafc53a0426f27650da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 16:32:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2cf29968b1134987b9cf1b255b4396a3\n\n- grenade https://zuul.opendev.org/t/openstack/build/86682f4769514246a5fc73e38bad3af8 : SUCCESS in 50m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ecc6ff84db6d4acd9a3f2c84840ad232 : SUCCESS in 1h 04m 21s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/0869eb4e37524c73a7682fc93ceba9fb : SUCCESS in 1h 57m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0cf95c1283d471894c4b2af2b1a65c4 : SUCCESS in 1h 07m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5448104ac0874028bd6d49797a7dee5b : SUCCESS in 11m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbd8fe0c05c443748aaa0ad52a5cda56 : FAILURE in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98b5ddfe3d934d6dbe882dcf30b3c065 : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/51abe3e062f745119734c74e9994814e : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0abb5ce9b0d247c4acac978a089677ba : SUCCESS in 7m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/97984112497542a88d274bcf656ad20a : SUCCESS in 6m 31s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0ffabe00adb14849a91d341ffe869351 : SUCCESS in 9m 06s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/84d53f826a7e4e86953af65dc0897158 : SUCCESS in 24m 07s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/4421b712f5ab482e81d2eb00400dffcc : SUCCESS in 22m 18s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/7c92611e427c4648b3941776f35f7145 : FAILURE in 3m 20s (non-voting)\n- tempest-integrated-placement-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18a5321a91f646c9bc82be95850b7e2f : SUCCESS in 1h 25m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed9bd77e1c0a43f0be05e302a5e705ab : SUCCESS in 36m 10s\n\nWarning:\n  Comments left for invalid file placement/tests/functional/test_allocation_candidates.py","accounts_in_message":[],"_revision_number":2},{"id":"e31de192750cc789e84065076f71016c3ab794bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-11 12:38:19.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"8f132297445bf3801de8a18ea5f357dc68af34ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 14:20:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbb611c846e2451883e861da6a56c816\n\n- grenade https://zuul.opendev.org/t/openstack/build/e5f3838b3ab34e8ba8fdb3078f321ea3 : SUCCESS in 36m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bd965b0f17d147228bcde0afa230c9e2 : SUCCESS in 50m 24s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/0f2e1cdc1fd54d3083202758363b3b5c : SUCCESS in 1h 36m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e7e3219dc16845ceb72b3be48f7d91ec : SUCCESS in 1h 03m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e440f2df6584f1d9ebad3fbd608a841 : SUCCESS in 14m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59ea2a3a6849454ba4a7e6f7a2372f5b : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8dcc913e147344468035e59d7b71dd1a : SUCCESS in 3m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c8d9617ab01f4a10b0dcf3523bb031c8 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e7395efa19a42d684d32fde73a71f18 : SUCCESS in 6m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1425ef6e179f45049d43e58c439e102d : SUCCESS in 7m 00s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4604d778d0d94751a3a84a9d6b1a30c3 : SUCCESS in 11m 01s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2fd968dc75614ed9b7e3963a4dbf5d36 : SUCCESS in 23m 51s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/6f53defd1bc44f8aac0f89cb09903f70 : SUCCESS in 22m 18s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/a088553d19ec447cad6a341093c2f3bc : FAILURE in 3m 24s (non-voting)\n- tempest-integrated-placement-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7fcc767dfa54a29a002e64d60a538f6 : SUCCESS in 1h 32m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/120f892514ee4f44ab3e3e29e5429661 : SUCCESS in 51m 34s","accounts_in_message":[],"_revision_number":3},{"id":"410dbcb93ee0e6b204b275377814a72f91deff3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-12 16:07:00.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"c085c03e6febacd253eaddf6e0d4e7c92988d157","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 18:10:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5c2f4e326bd4aef9873f6e59380e7af\n\n- grenade https://zuul.opendev.org/t/openstack/build/9bdb433187ad4e2ba131a7837b30b9ae : SUCCESS in 1h 01m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1551b728b3464e779d9cf167f2c9a969 : SUCCESS in 48m 28s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/6aaeaf81d8c84c6abfbd46e262f35343 : SUCCESS in 1h 43m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34b1a7ce7b8647bf8c18633753f41cb1 : SUCCESS in 1h 10m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/265c325f1e884c4f9a54b4491081b450 : SUCCESS in 10m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30c8f5744d694ed98fcb256d09250ce6 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0d8d328bf904233a02485895647f45a : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/71ad37aeedae4223b9b10658afee4d8f : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d361b1ce38bc4c56accc18e17fa03609 : SUCCESS in 7m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6743ea8a375748be96d8c8d8d821a174 : SUCCESS in 6m 33s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e5a249c3e3ac4ae092d142301c9c3443 : SUCCESS in 9m 15s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8a8a925a5567479aaeaab6846243e744 : SUCCESS in 24m 19s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/6c825c9385df45d1892b2f8be3ab4ee9 : SUCCESS in 16m 04s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/0c01893e5fad428c9bc81e32de3fe896 : FAILURE in 3m 04s (non-voting)\n- tempest-integrated-placement-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04a7df08e354408f9d29e3d1defeebed : SUCCESS in 1h 55m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1dec094216d4997b759e7fda58f69a3 : SUCCESS in 37m 29s","accounts_in_message":[],"_revision_number":4},{"id":"041f42d001e6e3fc68d980d976985dfe0234c1c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-09 08:29:52.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"7a14805826240e030e8c5fa9ec789088f7d4ad28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 10:13:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c05653189f241548da4b65651e064d9\n\n- grenade https://zuul.opendev.org/t/openstack/build/a4ff42d5409245c98029cd2cd2cef787 : SUCCESS in 1h 00m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e716950257bb47bd810380feb03506b6 : SUCCESS in 1h 02m 48s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/98739ad35ec84443aafd8be9c6c73290 : SUCCESS in 1h 26m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f5ae74be12c47bb922366b8faec7e5b : SUCCESS in 1h 00m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ac40ae8b85445f8bf32056544b0f050 : SUCCESS in 11m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9fa616d7755413c84838031099b95e5 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26901770aa704c6ba0faf4e881eadf22 : SUCCESS in 5m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/92be55ba2cbd4dd8844154496ee48d8f : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bef26c9ef2434ba6ab1e27c3e3f2183b : SUCCESS in 8m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/85a99e2fcc5549239ade4cdcbee1d0ce : SUCCESS in 6m 12s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/dbaf23c82b0c4db08f765fbfafc34ab4 : SUCCESS in 10m 40s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/27f7ab230d4649ac803a09ce02fea700 : SUCCESS in 24m 11s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/af766a3851934c0893e6cae612d15e00 : SUCCESS in 10m 46s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/cdf29aef387e4523abd4d35fcac23987 : FAILURE in 2m 53s (non-voting)\n- tempest-integrated-placement-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9005465bb6254ffea2aa585d14c10992 : SUCCESS in 1h 37m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dba9e45f3aee4ad5b81f0aee5f0be972 : SUCCESS in 54m 24s","accounts_in_message":[],"_revision_number":5},{"id":"069139eba6b7c945a4012fcd40619c2cf3d9cae0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-13 12:34:02.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d53895c5290c1a431db661868a9ed77e8bc9af4c","revisions":{"5291e14a41dca398d2e128c8bfd8e93da35a19af":{"kind":"REWORK","_number":1,"created":"2024-12-06 17:39:55.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/74/937274/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/74/937274/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/74/937274/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/74/937274/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/74/937274/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/74/937274/1"}}},"commit":{"parents":[{"commit":"49e58a3cc95b7200be5f798b21f91b752c898a15","subject":"Add round-robin candidate generation strategy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/49e58a3cc95b7200be5f798b21f91b752c898a15"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-06 17:39:26.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-06 17:39:26.000000000","tz":60},"subject":"DNM: test with breadth-first in tempest","message":"DNM: test with breadth-first in tempest\n\nChange-Id: I3218bbbecb6b2dd6e94fba270cf3084f278b404f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/5291e14a41dca398d2e128c8bfd8e93da35a19af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/5291e14a41dca398d2e128c8bfd8e93da35a19af"}]},"branch":"refs/heads/master"},"e85038a83ec885df30aa52b39e9b45868e60b6ba":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-12-09 13:40:38.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/74/937274/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/74/937274/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/74/937274/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/74/937274/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/74/937274/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/74/937274/2"}}},"commit":{"parents":[{"commit":"96f6ab5c21a243a4e8be7a2b45997e58591d9a67","subject":"Add round-robin candidate generation strategy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/96f6ab5c21a243a4e8be7a2b45997e58591d9a67"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-06 17:39:26.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-09 13:40:35.000000000","tz":60},"subject":"DNM: test with breadth-first in tempest","message":"DNM: test with breadth-first in tempest\n\nChange-Id: I3218bbbecb6b2dd6e94fba270cf3084f278b404f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e85038a83ec885df30aa52b39e9b45868e60b6ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e85038a83ec885df30aa52b39e9b45868e60b6ba"}]},"branch":"refs/heads/master"},"78b0c12b5154d1cb65c16faa370638a8ffe36d94":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-12-11 12:38:19.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/74/937274/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/74/937274/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/74/937274/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/74/937274/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/74/937274/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/74/937274/3"}}},"commit":{"parents":[{"commit":"7700465e28f701a09259261918e9464bf248f5bd","subject":"Add round-robin candidate generation strategy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/7700465e28f701a09259261918e9464bf248f5bd"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-06 17:39:26.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-11 12:38:06.000000000","tz":60},"subject":"DNM: test with breadth-first in tempest","message":"DNM: test with breadth-first in tempest\n\nChange-Id: I3218bbbecb6b2dd6e94fba270cf3084f278b404f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/78b0c12b5154d1cb65c16faa370638a8ffe36d94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/78b0c12b5154d1cb65c16faa370638a8ffe36d94"}]},"branch":"refs/heads/master"},"4c166ee5bb872261f2fc1ac32c00e0c9bb5f7d2f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-12-12 16:07:00.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/74/937274/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/74/937274/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/74/937274/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/74/937274/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/74/937274/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/74/937274/4"}}},"commit":{"parents":[{"commit":"e471db9574a3350e7e2038cbc3a10fd61559a186","subject":"Add round-robin candidate generation strategy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e471db9574a3350e7e2038cbc3a10fd61559a186"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-06 17:39:26.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-12 16:06:27.000000000","tz":60},"subject":"DNM: test with breadth-first in tempest","message":"DNM: test with breadth-first in tempest\n\nChange-Id: I3218bbbecb6b2dd6e94fba270cf3084f278b404f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/4c166ee5bb872261f2fc1ac32c00e0c9bb5f7d2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/4c166ee5bb872261f2fc1ac32c00e0c9bb5f7d2f"}]},"branch":"refs/heads/master"},"d53895c5290c1a431db661868a9ed77e8bc9af4c":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-01-09 08:29:52.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/74/937274/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/74/937274/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/74/937274/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/74/937274/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/74/937274/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/74/937274/5"}}},"commit":{"parents":[{"commit":"f20e13f0b233753a1b7f1899d24ad03d0d801b68","subject":"Add round-robin candidate generation strategy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f20e13f0b233753a1b7f1899d24ad03d0d801b68"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-06 17:39:26.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-01-09 08:29:07.000000000","tz":60},"subject":"DNM: test with breadth-first in tempest","message":"DNM: test with breadth-first in tempest\n\nChange-Id: I3218bbbecb6b2dd6e94fba270cf3084f278b404f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d53895c5290c1a431db661868a9ed77e8bc9af4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d53895c5290c1a431db661868a9ed77e8bc9af4c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
