)]}'
{"id":"openstack%2Fplacement~938940","triplet_id":"openstack%2Fplacement~stable%2F2024.2~I4f8c8491d21426310a82df7186b7689b2dad253e","project":"openstack/placement","branch":"stable/2024.2","topic":"bug/2070257","attention_set":{},"removed_from_attention_set":{"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2025-01-21 16:01:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4f8c8491d21426310a82df7186b7689b2dad253e","subject":"Factor out allocation candidate generation strategy","status":"MERGED","created":"2025-01-10 17:01:34.000000000","updated":"2025-01-21 16:02:18.000000000","submitted":"2025-01-21 16:01:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938940-bug/2070257","meta_rev_id":"61638bfff027904cb5c2af8948fae73289e95a4e","_number":938940,"virtual_id_number":938940,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-21 16:01:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2025-01-13 14:50:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2025-01-21 13:55:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2025-01-21 13:55:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-10 18:25:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-13 14:50:22.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-01-21 13:55:19.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"4d53c9ff0c7ec13840c1316f7e119f4349bee068","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-10 17:01:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"adc6f7674af9c84f993933964ba2071bfa8a95c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 18:25:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34f28f4529eb456fb84de9282d6fa901\n\n- grenade https://zuul.opendev.org/t/openstack/build/8793d8d2154a4f5b954b88bc3ffaea46 : SUCCESS in 1h 00m 01s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/d90650742a8543ddaec483f3c86d09f5 : SUCCESS in 1h 20m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d52cea6814b142a88453d13fe30ebd90 : SUCCESS in 38m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9b514109ede46e192b2c8df4c4f3764 : SUCCESS in 11m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe9c993d2b7f41dfb6925d08cf9fcbd0 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9cf75bd405ec408889659fad9e018824 : SUCCESS in 5m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/25bd6913dc05423a8c210a8e3d8436aa : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/461b2526d7214e37bf32b1a176cf1ecd : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe843f7ff0be4fccbbcebab5e211a33c : SUCCESS in 6m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f2fc27d343754ab7b57efb64789e0379 : SUCCESS in 7m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/feea621bd81941f09499e6b9e348671b : SUCCESS in 5m 22s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/636e0c3e1cfe48c8b5755ffa5804275b : SUCCESS in 20m 37s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/afe264175c104b7cab82b8fe1e6fafe7 : SUCCESS in 16m 56s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/3020821b975448a0bf2bd0ea90326cc2 : FAILURE in 5m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28142e01547444fabe3f32f7cc502453 : SUCCESS in 1h 05m 08s","accounts_in_message":[],"_revision_number":1},{"id":"76ceac615b4084b0148bd483c4a9388f6b0f1ca0","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-13 14:50:22.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"deb04756241146bafd2132740e2f084f8c0f9995","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-01-21 13:55:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1295f41bdee7d06a996d48d52fd7bb3c10330fc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 13:58:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"04aa2868310564b22ff42e6a17eccc5c19069543","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 16:01:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0566a881c8b24171b9f0ac81e8e07f2e\n\n- grenade https://zuul.opendev.org/t/openstack/build/6a9b5f8aa93b45f1b01c6fe9efa253e2 : SUCCESS in 1h 09m 51s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/211d852edc7c42ebbc6983acca1189c9 : SUCCESS in 1h 53m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e7612af1db93404cb7cd47ebf3c9b9b4 : SUCCESS in 1h 02m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b4a31639e6641efa1ec5ff87ff71d07 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3cc7f440f6ad4f598e26d7fdee2d1481 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/350760746c3f4813b30e1c69255e6434 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87041f7c2cc44857b207cb78c267904f : SUCCESS in 6m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/282fbd09c8f34b478086385a5cca31b3 : SUCCESS in 5m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c0aedd7836d94bd096724248c95c8a7c : SUCCESS in 6m 53s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/59eab3a278614933bee0635d98b4b993 : SUCCESS in 18m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/305946d79bb4459eae0729f9f32f6ddf : SUCCESS in 1h 03m 07s","accounts_in_message":[],"_revision_number":1},{"id":"bd70e928df0fbcfa2fc16fc025db2bc622253a3a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 16:01:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"61638bfff027904cb5c2af8948fae73289e95a4e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 16:02:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/851ba7e0fad149b99ad0a6ad088bdf15\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4c89294d5664c3296799388360daf18 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b0a927203ec779be631e3c61b9d6e4c7231cdec","revisions":{"1b0a927203ec779be631e3c61b9d6e4c7231cdec":{"kind":"REWORK","_number":1,"created":"2025-01-10 17:01:34.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/40/938940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/40/938940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/40/938940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/40/938940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/40/938940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/40/938940/1"}}},"commit":{"parents":[{"commit":"f276fc236383c87579cc1ba3bd6fe230570888fc","subject":"Add a global limit on the number of allocation candidates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f276fc236383c87579cc1ba3bd6fe230570888fc"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-06 15:47:53.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-01-10 17:00:48.000000000","tz":60},"subject":"Factor out allocation candidate generation strategy","message":"Factor out allocation candidate generation strategy\n\nThis is a pure refactor to pull out the logic of assembling an\nallocation candidate.\n\nThe _get_areq_list_generators returns a set of\ngenerators, one per root (anchoring) provider, that can generate all the\nviable candidates for that root by applying a product on the solutions\nfor each request groups.\n\nThe _generate_areq_lists implements the strategy of consuming these\ngenerators. The current behavior is that placement consumes the\ngenerator for the first root fully before moving to the next\n(depth-first).\n\nA later patch will change _generate_areq_lists to be able to apply\ndifferent strategies, like round-robin easily.\n\nRelated-Bug: #2070257\nChange-Id: I4f8c8491d21426310a82df7186b7689b2dad253e\n(cherry picked from commit 04718d65b60e817b5f47c5a3f50e302fe2302dfd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/1b0a927203ec779be631e3c61b9d6e4c7231cdec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/1b0a927203ec779be631e3c61b9d6e4c7231cdec"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
