)]}'
{"id":"openstack%2Fplacement~938942","triplet_id":"openstack%2Fplacement~stable%2F2024.1~I5475c66fa35958b3dc72c3bab7a7edbdf6c89ce6","project":"openstack/placement","branch":"stable/2024.1","topic":"bug/2070257","attention_set":{},"removed_from_attention_set":{"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2025-04-10 18:37:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5475c66fa35958b3dc72c3bab7a7edbdf6c89ce6","subject":"Add a global limit on the number of allocation candidates","status":"MERGED","created":"2025-01-10 17:03:03.000000000","updated":"2025-04-10 18:38:24.000000000","submitted":"2025-04-10 18:37:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938942-bug/2070257","meta_rev_id":"aafdb8bbf4ec1cfeaeb9b908c3d3cb6c19a6fd17","_number":938942,"virtual_id_number":938942,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-10 18:37:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-02-03 16:49:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2025-02-27 12:35:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2025-02-27 12:35:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-10 18:45:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-03 16:49:25.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2025-02-27 12:35:49.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"249d58b569382456c5e395a5993cef03fb6d9878","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-10 17:03:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c9ad214f60bf9b5c9dfa294b555c2dae65e79f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 18:45:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ba10789b8dc4ba0b4698fa51eab3f30\n\n- grenade https://zuul.opendev.org/t/openstack/build/aaa6726f493446e6adf3fd1284ce6f95 : SUCCESS in 1h 09m 07s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/bceab327ad294c3d8dc83288f980e88a : SUCCESS in 1h 13m 53s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/e22d7d9d95f74644b288e9da7f9ea646 : SUCCESS in 1h 35m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72b3bf79653d46f3b69f55fa1e47a6f0 : SUCCESS in 1h 05m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba1b2002e49840baade0c0896dcd5258 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d2a55e9fbf7414287a5ae4bc7d52017 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c0ba61fe71f4cd582760726cc940320 : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f29adf2e11094357b40782daf891a364 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98f96d41f591402c82faefb870d77142 : SUCCESS in 5m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d59ebb8215c9461e9c2269278ab588ea : SUCCESS in 6m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/eaa5c0d95058434aa4c7edbeb4dcf0de : SUCCESS in 6m 31s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/89bfa553e8f64a40ba2221f52916aa33 : SUCCESS in 19m 11s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/f1b3be1259db41ebabcb5490c4f70477 : SUCCESS in 20m 49s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/b61b755fd10e4fd3886d44844d08ed81 : FAILURE in 3m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0639795b2c5f498e88f41d30426bbceb : SUCCESS in 54m 03s","accounts_in_message":[],"_revision_number":1},{"id":"32e3a27c14476d61bf3c60847cc1d5fa710483e9","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-02-03 16:49:25.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"57e131a621f382f9bcf9ac720aad076cbbbc6444","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-02-27 12:35:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c0f76de9ad1394e4559794e0f2c10c5b1f0e07d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 12:36:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c6023414e50e9f6f517ac7ae6943438cc3ce6aeb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 01:04:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/82167d646fe14ad0b850471106b87e0c\n\n- grenade https://zuul.opendev.org/t/openstack/build/a943a5b20a17406fa6f9218181393bb8 : SUCCESS in 1h 16m 00s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/13b5a95bdc3d4dbe85a18bf527100ba0 : SUCCESS in 1h 53m 00s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/141d40defa4d49dd829faceb8e8a9c46 : SUCCESS in 1h 03m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/360525c8ec954a93a35aee77abfcab97 : POST_FAILURE in 20m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dc794bcbd1e484193f37026ddca9cbd : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/785036e9238b4db5b9764c8a09abce08 : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f13ec4d0224c440c955596ed61c2693f : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c029c0c1bc94199b22c04ba74874099 : SUCCESS in 8m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a22fa364a8d44ba7a208c37ab305c9ab : SUCCESS in 6m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3666460e2d734d09a1396802e858d3cf : SUCCESS in 11m 31s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/844024966de44a909a8d1ccb9a4a864a : SUCCESS in 18m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ebd118b658824a5284b5bc998b5e65c2 : SUCCESS in 1h 09m 27s","accounts_in_message":[],"_revision_number":1},{"id":"f23752d65e273755f9946d8cb7a2d2d1b767cdae","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-04-10 15:06:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cb80f5daf0e7bfa1104634fc61ff8d81de1965d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 16:59:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7697545fd88849ca94f27e9867d71377\n\n- grenade https://zuul.opendev.org/t/openstack/build/de1b288c29b84b28b24420ea48abedba : SUCCESS in 49m 10s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/839ed08d32a5478a97e10fb0ba6e80c6 : SUCCESS in 58m 11s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/e14573a65d5f4c919e4e31851e135020 : SUCCESS in 1h 43m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2337916b5a49411980d2a4e3c7b8dd50 : SUCCESS in 56m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c85d32d4b51430ea8ef827f9bb975d7 : SUCCESS in 8m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49a8438eccdd40f4be11e3ce015bd0a9 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/580de7cb45de4aba9160faeabaef21ba : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d10a630de2df4187a18e4695341f896a : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f28d36880b03449c8be29b3caa6da084 : SUCCESS in 6m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/81bf095868b5479087f6f68b892181ba : SUCCESS in 6m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fe9344acceb34104853b838ae262b452 : SUCCESS in 7m 08s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5d6a4b56066d423090e99756099caf45 : SUCCESS in 15m 15s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/7ed2e4878e0f442da529b77e5a22a35e : SUCCESS in 21m 45s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/b99d597ac0574552a843ec67dfd9a5bf : FAILURE in 2m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02ad9d461cff41fe98cf8d1a99d1bff9 : SUCCESS in 55m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a96c013939d568aaabf4cc6641ee5a9a1f3b370a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 16:59:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d5d8eeca37f42e7f6b8806c5882e2176850c2d6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 18:37:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ffbab9485bb48df9c05520047107bf9\n\n- grenade https://zuul.opendev.org/t/openstack/build/96d20af91a2f4a1c9aac9c4a1705a6c8 : SUCCESS in 49m 15s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/328ec0a5561845fbb4bc677cbf7aac39 : SUCCESS in 1h 33m 24s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9155132c3d9e4da199c1e6e2df96af81 : SUCCESS in 49m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b80caff6619e4c0686c70f711ff42c77 : SUCCESS in 58m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1c609f181964d97925db2d8f20e1c2c : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99bb0a53ce874bb48d9221a3ea75b2f8 : SUCCESS in 5m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d7fb9ddf082a44ec92c3bc834dc6935e : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/765ce70c0b3e4f078ad2f29904a6bfa6 : SUCCESS in 6m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b0e751163b1b402086aa4bffb61a0dfb : SUCCESS in 7m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/44f66fa8174d4da3b6a3b4db8f78569e : SUCCESS in 6m 06s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/94188f3e5d37400eb66343b7e2be1dce : SUCCESS in 16m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a748dd656bf34895b4eebbd8d8d28365 : SUCCESS in 41m 22s","accounts_in_message":[],"_revision_number":1},{"id":"9695afa0eea0ab2d86e1982ef7486be053d76a24","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 18:37:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"aafdb8bbf4ec1cfeaeb9b908c3d3cb6c19a6fd17","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 18:38:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4080130fb2b40d69bb13bfdceb75d9a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcd10bef9e25493eb5ba40a7e62f3029 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"47ee16bdc83e9959c6f617be8fd93383394ca2bd","revisions":{"47ee16bdc83e9959c6f617be8fd93383394ca2bd":{"kind":"REWORK","_number":1,"created":"2025-01-10 17:03:03.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/42/938942/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/42/938942/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/42/938942/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/42/938942/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/42/938942/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/42/938942/1"}}},"commit":{"parents":[{"commit":"af00c864e98ae76cae3e621ad95868234d027882","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/af00c864e98ae76cae3e621ad95868234d027882"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-11-28 19:21:24.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-01-10 17:02:38.000000000","tz":60},"subject":"Add a global limit on the number of allocation candidates","message":"Add a global limit on the number of allocation candidates\n\nWhen a root provider has multiple similar child RPs, i.e. we have a wide\nbut symmetric tree, and the allocation\ncandidate request asks for resources in multiple request groups the\nnumber of possible allocation candidates grows rapidly.\n\n* 1 root, 8 child RPs with 1 unit of resource each\n  a_c requests 6 groups with 1 unit of resource each\n  \u003d\u003e 8*7*6*5*4*3\u003d20160 possible candidates\n\n* 1 root, 8 child RPs with 6 unit of resources each\n  a_c requests 6 groups with 6 unit of resources each\n  \u003d\u003e 8^6\u003d262144 possible candidates\n\nPlacement generates these candidates fully before applying the limit\nparameter provided in the query to be able do a random sampling if\n[placement]randomize_allocation_candidates is True.\n\nPlacement takes excessive time and memory to generates these amount of\nallocation candidates.\n\nTo avoid request timeout or out of memory events this patch introduces a\nnew [placement]max_allocation_candidates config option that is applied\nnot after but *during* the candidate generation loop. So this new option\nacts as a global safety limit.\n\nThe new config option is defaulted to -1, meaning no limit, to keep the\nlegacy behavior, but we suggest to tune that config in environments with\nwide symmetric trees.\n\nPlacement generates the candidates with a depth first approach,\ngathering all candidates for a root before moving to the next root. If\nthe new global limit is applied and hit then it is possible that the\nreturned candidates are from a limited set of root providers. This issue\nwill be handled by a subsequent patch allowing to select from\ndepth-first or breadth-first (round-robin) candidate generation\nstrategy.\n\nRelated-Bug: #2070257\nChange-Id: I5475c66fa35958b3dc72c3bab7a7edbdf6c89ce6\n(cherry picked from commit 9351241f91f80cb7dc45c4c0daab929b5dfc5c4b)\n(cherry picked from commit f276fc236383c87579cc1ba3bd6fe230570888fc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/47ee16bdc83e9959c6f617be8fd93383394ca2bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/47ee16bdc83e9959c6f617be8fd93383394ca2bd"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
