)]}'
{"id":"openstack%2Fplacement~938945","triplet_id":"openstack%2Fplacement~stable%2F2023.2~I5475c66fa35958b3dc72c3bab7a7edbdf6c89ce6","project":"openstack/placement","branch":"stable/2023.2","topic":"bug/2070257","attention_set":{},"removed_from_attention_set":{"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2025-04-11 11:45:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5475c66fa35958b3dc72c3bab7a7edbdf6c89ce6","subject":"Add a global limit on the number of allocation candidates","status":"MERGED","created":"2025-01-10 17:04:14.000000000","updated":"2025-04-11 11:46:30.000000000","submitted":"2025-04-11 11:45:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938945-bug/2070257","meta_rev_id":"262c73fa51ac743e2cc8f3d976cb99b863354147","_number":938945,"virtual_id_number":938945,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-11 11:45:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-04-11 10:19:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2025-02-27 13:14:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-04-11 10:19:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-10 18:19:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-27 13:14:37.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2025-04-11 10:19:12.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"bb47982a014265a8b4799597e4ee8dddbe8fc3aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-10 17:04:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12cf698ccfc7b368972a2306dc79cc17b3f40d72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 18:19:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94a2c088c0fd4e639c1ca4d1c0c7fcca\n\n- grenade https://zuul.opendev.org/t/openstack/build/4aa24cf0beef4344b29c12c7cc0cb404 : SUCCESS in 44m 42s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/09b39b576e87483c835e33c0cbffb738 : SUCCESS in 1h 09m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e8795fb0a1a47129b9ca7700addee80 : SUCCESS in 44m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a29949207e54a1e8fef56cc2af6572e : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a873da0f511b4e9b8607c031eaba7c96 : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1bc148a9b7744dbcac96959d8813486c : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3168386c806345cb97ca65e190b2b007 : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2a1a4d193a24f129c7efc1219591d17 : SUCCESS in 5m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5d3dca9791b34b1ca7a258b2d5518e93 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8719e4846f342909dc73480b72f2ba9 : SUCCESS in 8m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e8df6d70115f40e0baae977cdf17302a : SUCCESS in 5m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/40a899790f3743948c8587b58b1537e1 : SUCCESS in 6m 26s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cd78751e86824f728582c04034d71c11 : SUCCESS in 19m 03s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/1f9c2f77eb0c4e21b8d48950ef46cd5a : FAILURE in 4m 57s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/ae1d82cc17654c89adde312d9dc86b08 : FAILURE in 3m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ddcc6748ac054f2686760a2daed4c2fc : SUCCESS in 1h 08m 01s","accounts_in_message":[],"_revision_number":1},{"id":"8758d98be3f2dd9d39511cba28ccd97734ef61ea","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-02-27 13:14:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0f4239de39035c6954e11950c2980ffc6915f866","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-04-11 10:19:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c253ca48f8123ee48c944e5b0d5490a41c6ffe75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 10:19:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3283199518ba68b8716392f2a3a213e7b7ced663","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 11:45:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d04faa742b484914b7496f7bdd6ddb9c\n\n- grenade https://zuul.opendev.org/t/openstack/build/8633887cf77c4f53840c129660391df4 : SUCCESS in 51m 14s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/8813ea8342024f0fa675de3276a00720 : SUCCESS in 1h 21m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/094f89a3f9564afba4b09e4a494ea05e : SUCCESS in 50m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dee064bff6f4a0e987f0c0ca1027065 : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a1816b67ec74b3eb3a784695ffcf5a5 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cdfc17ed4f8e40a0accea779196e5dba : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/93bd4b9cf26b45079e94b58ab84a3490 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16f074f92015437b92da1f6f33e72870 : SUCCESS in 7m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1b77db9d076442f584c2ffdfc2a606da : SUCCESS in 3m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d4d1b435c8694a18b9689ffe19e91c26 : SUCCESS in 6m 40s\n- placement-nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7bb2c0099c0f4699999eb38aef137153 : SUCCESS in 24m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6d6c833b5b54ee1969c805da80e3a84 : SUCCESS in 23m 59s","accounts_in_message":[],"_revision_number":1},{"id":"24dfcb5a473e83012d6d13d5c2d9713e064cf5d4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 11:45:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"262c73fa51ac743e2cc8f3d976cb99b863354147","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 11:46:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff000050e3d34152acc84c7b90ecd793\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf0989871992444ab1be560493e82d92 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a1db67cdeb8fdc4361d2d8ec6097dc7bfd60e0ba","revisions":{"a1db67cdeb8fdc4361d2d8ec6097dc7bfd60e0ba":{"kind":"REWORK","_number":1,"created":"2025-01-10 17:04:14.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/45/938945/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/45/938945/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/45/938945/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/45/938945/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/45/938945/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/45/938945/1"}}},"commit":{"parents":[{"commit":"a361622d749d3b24aad638ec1b03a7d7124a87b3","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/a361622d749d3b24aad638ec1b03a7d7124a87b3"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-11-28 19:21:24.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-01-10 17:03:49.000000000","tz":60},"subject":"Add a global limit on the number of allocation candidates","message":"Add a global limit on the number of allocation candidates\n\nWhen a root provider has multiple similar child RPs, i.e. we have a wide\nbut symmetric tree, and the allocation\ncandidate request asks for resources in multiple request groups the\nnumber of possible allocation candidates grows rapidly.\n\n* 1 root, 8 child RPs with 1 unit of resource each\n  a_c requests 6 groups with 1 unit of resource each\n  \u003d\u003e 8*7*6*5*4*3\u003d20160 possible candidates\n\n* 1 root, 8 child RPs with 6 unit of resources each\n  a_c requests 6 groups with 6 unit of resources each\n  \u003d\u003e 8^6\u003d262144 possible candidates\n\nPlacement generates these candidates fully before applying the limit\nparameter provided in the query to be able do a random sampling if\n[placement]randomize_allocation_candidates is True.\n\nPlacement takes excessive time and memory to generates these amount of\nallocation candidates.\n\nTo avoid request timeout or out of memory events this patch introduces a\nnew [placement]max_allocation_candidates config option that is applied\nnot after but *during* the candidate generation loop. So this new option\nacts as a global safety limit.\n\nThe new config option is defaulted to -1, meaning no limit, to keep the\nlegacy behavior, but we suggest to tune that config in environments with\nwide symmetric trees.\n\nPlacement generates the candidates with a depth first approach,\ngathering all candidates for a root before moving to the next root. If\nthe new global limit is applied and hit then it is possible that the\nreturned candidates are from a limited set of root providers. This issue\nwill be handled by a subsequent patch allowing to select from\ndepth-first or breadth-first (round-robin) candidate generation\nstrategy.\n\nRelated-Bug: #2070257\nChange-Id: I5475c66fa35958b3dc72c3bab7a7edbdf6c89ce6\n(cherry picked from commit 9351241f91f80cb7dc45c4c0daab929b5dfc5c4b)\n(cherry picked from commit f276fc236383c87579cc1ba3bd6fe230570888fc)\n(cherry picked from commit 47ee16bdc83e9959c6f617be8fd93383394ca2bd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/a1db67cdeb8fdc4361d2d8ec6097dc7bfd60e0ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/a1db67cdeb8fdc4361d2d8ec6097dc7bfd60e0ba"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
