)]}'
{"id":"openstack%2Fplacement~945464","triplet_id":"openstack%2Fplacement~master~If59d2c54c3e32fda285ab7d8f5366393a3f69481","project":"openstack/placement","branch":"master","topic":"bug/2104040","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-08-27 14:31:37.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2025-08-29 10:04:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If59d2c54c3e32fda285ab7d8f5366393a3f69481","subject":"Reproduce bug 2104040: allocate in over-capacity","status":"MERGED","created":"2025-03-25 14:34:44.000000000","updated":"2025-08-29 10:06:50.000000000","submitted":"2025-08-29 10:04:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945464-bug/2104040","meta_rev_id":"ed302e87c6e19b9cc7355ed8c6d810bb7bfbe4f1","_number":945464,"virtual_id_number":945464,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-29 10:04:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2025-08-27 14:31:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2025-08-27 11:48:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2025-08-27 14:32:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-20 13:21:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2025-08-27 14:32:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-25 15:32:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-24 17:42:11.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-05-20 13:21:08.000000000","updated_by":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"reviewer":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2025-08-27 11:48:16.000000000","updated_by":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"reviewer":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"state":"REVIEWER"},{"updated":"2025-08-27 14:32:11.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"68c9933c2229668683f528fb2563c713200f64b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-03-25 14:34:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d441e642b74a7f7126fa7721a82649273dd9b59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 15:32:38.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/379960ba5a114a96bff411b6d2cce67f\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e58026eef3ef4ebea3a339b8a7ccd36c : SUCCESS in 43m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1deea70395f846db8c5e9b8a557f1268 : SUCCESS in 56m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8fe706a6b15848ecbe29fd4e0f3e8af6 : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3382561dff604a3b84f1e694d59219d9 : FAILURE in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/679574337dc2463bba069353a0f1cd82 : SUCCESS in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b97b20f0f1d14da2986dda29c1b150d7 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49265127a06a422ea0dadf15a5c99148 : SUCCESS in 11m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e022d60cfca748ddab487818d4f55f68 : SUCCESS in 7m 17s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/956c0e4b414e4f5fbc8ab85e16511fa0 : SUCCESS in 11m 00s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/eb76be38831f47c98f1d15fc8f77a130 : SUCCESS in 21m 50s","accounts_in_message":[],"_revision_number":1},{"id":"f7ced4059240cf7fd2f04ffab52cee068c04b65d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-03-25 15:40:42.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ba77735eed0fc278aabb04b2891312d9701c9225","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 16:49:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a15024a354594820bc6b370c12cdf966\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eb7d6ebaaa0d4edbbe84735600ef58b8 : SUCCESS in 53m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1da7613fcb0a4ea4b98ec2ddedf309f2 : SUCCESS in 53m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3cdd8dd352e2425b967240f53f8516d2 : SUCCESS in 17m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/166401e45bbf485d824199003dabda97 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ea01eacd3ec4f7da497ebe205ff4801 : SUCCESS in 6m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/43a26582db0f493cac6655051de6a0fc : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4436d0d665f642bdb1ec975eeb7a2e22 : SUCCESS in 8m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/35558856096548b984e5900ee9785188 : SUCCESS in 7m 09s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b652f163b95742b7ad3b409c5b302308 : SUCCESS in 9m 35s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/91b4ddf680274c55af8af7d0a35edc77 : SUCCESS in 26m 03s","accounts_in_message":[],"_revision_number":2},{"id":"33f9d14590ac768aff54d434d205bbc98b97c16b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-24 17:42:11.000000000","message":"Patch Set 2: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2dffbb0afad1d9381317becb7c0be1e72bc5e391","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-04-24 18:59:56.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"199a515a5ed02a744e2f1677dcc78763c0ae1860","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-04-24 19:10:17.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"be68879aa95485c128743d6fa48a44112616ccf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-04-24 19:10:24.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"06609f593200f3890cecf0d4ae14aee02f14a57a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 20:17:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c561f4363e048abbad61bab1c0db1ae\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/84ae67e846484b4fbcc75f8a20d24579 : SUCCESS in 55m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1842e3f1c0ed4f34902d78c72650f31e : SUCCESS in 1h 02m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35e9afc796fe47a5bc213b414b214f0a : SUCCESS in 13m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65577bfc7ff4403083fb4e881af782dc : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9bccba1c2e9e43788edd03ff465a27a1 : SUCCESS in 3m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e2714256172e468e860822ebc3837f47 : SUCCESS in 5m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5d18ff55887a4350ab4e6d71024d0d96 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e965f5a98844eee912c3ce832e2af7b : SUCCESS in 7m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9a696ec4bc4c44099db90a725b57e0a8 : SUCCESS in 6m 51s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/386bb36c76e0434dafcfa745e3d0b2cf : SUCCESS in 8m 09s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/15b62b6fd34f4120b4122ce80a30b14c : SUCCESS in 19m 49s","accounts_in_message":[],"_revision_number":3},{"id":"e1e9fa6e36a181d87a78055abba894f02d44992f","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-05-20 13:21:08.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ba0deb8bf60ce57134693de2833de50d72bff44a","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2025-08-27 11:48:16.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d6a2c945897cb6c925b59060d38af469f8d35319","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-08-27 14:31:37.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"118c01ad3567f2153ed7e55dd03dcf2921fbcecc","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-08-27 14:32:11.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ade4d0a94c94f2bff2c7cd3e2f17af97647af6a1","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-08-27 14:32:19.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e0c2a5210c2d521faa2c8f67d968b0b134ca45c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 14:33:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ab0afdff64b1f7c57315f5d8ec15b1e44aa30005","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 17:57:13.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b2ebbe8a27cc46f6852ade1e9f44f666\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1a71495bad174e41828168e8b9f0e0e0 : SUCCESS in 38m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7060b40072254694a676f0ae928ea7a5 : SUCCESS in 34m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7454028cf5b946ebb2215b171b2e0e22 : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/082a9e28ea2345d29710bd2e959d15e3 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f5880a3355bd40d682133596e9d215ed : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8117924756fe4b00a6b87cb70670a6c2 : SUCCESS in 5m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9539160a65bf465b89581f44c34a3ac2 : FAILURE in 6m 30s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/325aa92c74344343a191a8cc7721dd44 : SUCCESS in 5m 33s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e1eb67a51384489ab03a5950133e1aae : SUCCESS in 19m 39s","accounts_in_message":[],"_revision_number":3},{"id":"afaf6d8c1bc0d0b3901881571ae92ab2b306cc6c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-08-27 18:29:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2075816d3b6bc31b667d3d60d916c01e1d57c0f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 23:13:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/12ba0509e03846b89b7539e37118f5b4\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/15f2aecc42c649cbab0448501e1d2215 : SUCCESS in 34m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/728e33be4dc44283834f5ff4cf2f12f3 : SUCCESS in 32m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32589c2c6cd04b35aa9d6077cfffa79e : SUCCESS in 49m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a5cdee8cb014ef4a188027b3b62cc0e : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c777c0c67ff040f8800a25bd9a00874a : SUCCESS in 37m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6fea4b1de332497c9c9301b0501fe0ad : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/13beac9b46e54fc09d4a75c735c272d8 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3c0183599ed4935baa86da020edc351 : SUCCESS in 4m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fca19ae029a2402ea032b82f0ee6ad80 : FAILURE in 2m 50s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/74e3a4aff45940a4b494caa5df5e6163 : SUCCESS in 10m 51s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e5c7e5c0fe75441ab513cd66646e163e : ERROR Failed to update project openstack/nova in 4h 39m 50s","accounts_in_message":[],"_revision_number":3},{"id":"54a18538978c795ddfc4fd6e2dee93533cd2254a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-08-28 13:39:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6a192e13366f8d655fa5f6aec9ef462ee8c1742f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 14:54:30.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e3b63898b784a37bf8a46b8edd50625\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/40caf60a953143b48bfd8afaa1c0c64f : SUCCESS in 1h 07m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2f1c8f83adf47f4bd5b87d10bd50687 : SUCCESS in 58m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1030226034584b96b0d1400154969a15 : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/558e33e5943442e9840af07f5718fc99 : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/abd4e081ac0742bf8d9c4dd80d116fef : SUCCESS in 5m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3942d38b75574c10b55393c61aa23831 : SUCCESS in 5m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fb8d7eed63894ce38ff3a21dc9e54d39 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08e11234ce664fda944ec97c737c29be : SUCCESS in 9m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5af977a5805240ce858b62eb00506043 : FAILURE in 4m 44s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9fc5d4409a6d4d6f8b5982b1f459716a : SUCCESS in 24m 55s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d593263be9104cec845ffa7a0fb4b442 : SUCCESS in 24m 32s","accounts_in_message":[],"_revision_number":3},{"id":"54609c1bc4778a06828ea8452b1525a3d88947c1","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-08-29 07:28:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f00816c2a12f8f9b32a4f4ec2d899015ecedbadd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 08:51:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f45b0e08941f4db4b1be228199c4efb2\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5e1fd641e0ec445b82bd01dc358eb93a : SUCCESS in 29m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ef99eba4976413bb534b6ad7adaa31d : SUCCESS in 1h 16m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/097ce86955654c78822e2e48c5517485 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a82c1a8ab30a408c9b1e946954fa96f9 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dda1059513f949e494d213f9d9345294 : SUCCESS in 3m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2beae95062d64bf29cf7b493293624fe : SUCCESS in 7m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/524470d7060d48a0bff068fa419b4e82 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d901bb8e535a4d978d840c49f305503b : SUCCESS in 10m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/33ff7050e118476fb0e81c60695a96d1 : SUCCESS in 6m 02s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/7560502bd8e547ba8e357f4c4835afdc : SUCCESS in 4m 59s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/641f1f1936ab43f289e0eb4a510d0bff : SUCCESS in 21m 22s","accounts_in_message":[],"_revision_number":3},{"id":"a85e0b40e1cd86ba2b9e9382cde979286a5186dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 08:52:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"45b1fef4dc6ad25295a170967f1a2f1f54a2ac10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 10:04:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce6ee2ea76e24902b402137c815a5965\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d53f13d6d06e4962ae7932cd0eb4764a : SUCCESS in 57m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a2feb3a95414dbf8a0618d4ea993a03 : SUCCESS in 53m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85954e2a813147b0906c518b768234b0 : SUCCESS in 5m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2400f07f0e334dfcae156caae15152ae : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3120bbd04eaf4c94b7093dd2b6065fdd : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d25eac5b2d994ce9b84b15b3d1c2f9e3 : SUCCESS in 9m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2a1a0db016344e0298e3003053a89d54 : SUCCESS in 7m 20s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/71e9556b3d6f451985f59615c2d13a68 : SUCCESS in 5m 07s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c21003c03bc240babbd94d02e76e8390 : SUCCESS in 26m 32s","accounts_in_message":[],"_revision_number":3},{"id":"30f7eecb43fbb0fbee46de348cc50043a420ba62","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 10:04:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ed302e87c6e19b9cc7355ed8c6d810bb7bfbe4f1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 10:06:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71369b34594c402cac774d327e0bb770\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a279beb799d40c88b9dbfe8c8847804 : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"356ff48bb8d8b157d13e806b3003e062d6911933","revisions":{"90dd48feed9885717e8d065986ae7078cce9fa8d":{"kind":"REWORK","_number":1,"created":"2025-03-25 14:34:44.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/64/945464/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/64/945464/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/64/945464/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/64/945464/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/64/945464/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/64/945464/1"}}},"commit":{"parents":[{"commit":"b84329b3f767e284db75c370de3b7b71f688bb28","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b84329b3f767e284db75c370de3b7b71f688bb28"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-03-24 17:53:57.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-03-25 13:38:47.000000000","tz":-420},"subject":"Reproduce bug 2104040: allocate in over-capacity","message":"Reproduce bug 2104040: allocate in over-capacity\n\nThis reproduces the bug where a provider that is thrust into an over-\ncapacity situation has few options for getting out of that jam.\n\nChange-Id: If59d2c54c3e32fda285ab7d8f5366393a3f69481\nRelated-Bug: # 2104040\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/90dd48feed9885717e8d065986ae7078cce9fa8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/90dd48feed9885717e8d065986ae7078cce9fa8d"}]},"branch":"refs/heads/master"},"2473ebd30926c9ab44013068332154b47d09598f":{"kind":"REWORK","_number":2,"created":"2025-03-25 15:40:42.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/64/945464/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/64/945464/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/64/945464/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/64/945464/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/64/945464/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/64/945464/2"}}},"commit":{"parents":[{"commit":"b84329b3f767e284db75c370de3b7b71f688bb28","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b84329b3f767e284db75c370de3b7b71f688bb28"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-03-24 17:53:57.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-03-25 15:38:08.000000000","tz":-420},"subject":"Reproduce bug 2104040: allocate in over-capacity","message":"Reproduce bug 2104040: allocate in over-capacity\n\nThis reproduces the bug where a provider that is thrust into an over-\ncapacity situation has few options for getting out of that jam.\n\nChange-Id: If59d2c54c3e32fda285ab7d8f5366393a3f69481\nRelated-Bug: # 2104040\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/2473ebd30926c9ab44013068332154b47d09598f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/2473ebd30926c9ab44013068332154b47d09598f"}]},"branch":"refs/heads/master"},"356ff48bb8d8b157d13e806b3003e062d6911933":{"kind":"REWORK","_number":3,"created":"2025-04-24 19:10:24.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/64/945464/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/64/945464/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/64/945464/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/64/945464/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/64/945464/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/64/945464/3"}}},"commit":{"parents":[{"commit":"b84329b3f767e284db75c370de3b7b71f688bb28","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b84329b3f767e284db75c370de3b7b71f688bb28"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-03-24 17:53:57.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-04-24 19:07:57.000000000","tz":-420},"subject":"Reproduce bug 2104040: allocate in over-capacity","message":"Reproduce bug 2104040: allocate in over-capacity\n\nThis reproduces the bug where a provider that is thrust into an over-\ncapacity situation has few options for getting out of that jam.\n\nChange-Id: If59d2c54c3e32fda285ab7d8f5366393a3f69481\nRelated-Bug: #2104040\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/356ff48bb8d8b157d13e806b3003e062d6911933"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/356ff48bb8d8b157d13e806b3003e062d6911933"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
