)]}'
{"id":"openstack%2Fplacement~947483","triplet_id":"openstack%2Fplacement~master~Icce0fbb080f5b4d5022be8eb82251d916d544fdd","project":"openstack/placement","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-11-19 11:53:09.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-11-19 13:54:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icce0fbb080f5b4d5022be8eb82251d916d544fdd","subject":"Replace UPPER_CONSTRAINTS_FILE","status":"MERGED","created":"2025-04-16 13:31:24.000000000","updated":"2025-11-19 13:56:03.000000000","submitted":"2025-11-19 13:54:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947483","meta_rev_id":"4ee7b29f76f986f379232289082dcc6a3268d1ee","_number":947483,"virtual_id_number":947483,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-19 13:54:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-11-19 11:53:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-17 13:29:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-11-19 11:53:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-16 14:22:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-17 13:29:40.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-04-17 13:29:40.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"466ebdbee321c6af5079057242c03fca7d3388ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-16 13:31:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2ba8509e8e84f169e84f6006862fcc617025f17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 14:22:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f949590de41e4e0a83a086e2da368d01\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/44ed3db5ae8f4206bc14ddb19be4c27f : SUCCESS in 6m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e0428ea391604d1b824677d19aa26059 : SUCCESS in 49m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/871e09905f8a4e9f9a9c4f7ddf0294f0 : SUCCESS in 49m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/666e4d9f5d5d4be5afdd2f927b42c612 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63d261a3302f461cb9f73ed7add0b59e : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/41dd73fc973c44b68711a446b6e158ec : SUCCESS in 2m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b5e831be37f1401a8eccde650b6ba472 : SUCCESS in 5m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1ae80e2d8c7a4c299745dbf846b2260b : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60ac85d5370d422495cbabcdda99dbc0 : SUCCESS in 9m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c2350b813e04a94954cb29d097d5e1e : SUCCESS in 4m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/197e72e0ec094ad992433d2ebd27ca31 : SUCCESS in 3m 58s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8e66b42be48c4a798293c2452102146a : SUCCESS in 5m 52s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/90154013b99a46918a4e04638f40a447 : SUCCESS in 22m 52s","accounts_in_message":[],"_revision_number":1},{"id":"59459d6fecd0c5e35aa8ae420485062ebae90755","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-17 13:29:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8ff6ffebca5d0c1c9c025606321ecdffc9e14804","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-19 11:53:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7939ddb73a0d70c062c34dc23de7df16b00df6fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 11:53:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"190400b75758ef88447416fdc92ed5036a1f0107","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 13:54:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f578e4ca5854e29a0745df6c372eba3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b12fb7a74f8747548443e0b91f3d1ed8 : SUCCESS in 4m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ecebbde8b39f4537844e15741d09a878 : SUCCESS in 57m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7fd0924ea2454f3085ce5a5fc7645460 : SUCCESS in 1h 05m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a9197e63ca24b8c8fb8cc4a486e1dc3 : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca2e158c5037493799e54b382da5effb : SUCCESS in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ea28b56d7eac407ca6b68d373a93c8ab : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f76a94c1c56e4a26b2c3b2a53a524d04 : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/215a65df48064f6f81bd0a76a8672fbe : SUCCESS in 4m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fbb0a52bad0b4487a1fa92fae4c69c4c : SUCCESS in 3m 35s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b53cf13be1ad46d1970f6c901a7dfc90 : SUCCESS in 9m 34s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/386c26708d0249f4aba2b9f47f88a9ae : SUCCESS in 19m 47s","accounts_in_message":[],"_revision_number":1},{"id":"8b4b9dddadca628e34a7735625a862da6128f486","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 13:54:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4ee7b29f76f986f379232289082dcc6a3268d1ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 13:56:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/479d18c4a4fd4ecb818d5197cc3cde81\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/037a51783ab34cc1a3bd374c48083fa2 : SUCCESS in 40s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9931bee89e9423f865c9c97aca21ef3 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c245f6ac18e343ee89e92faa6f0b2d73 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"20e3f52bbcdc2fecef0bd6d88bc68ab26a27098b","revisions":{"20e3f52bbcdc2fecef0bd6d88bc68ab26a27098b":{"kind":"REWORK","_number":1,"created":"2025-04-16 13:31:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/947483/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/83/947483/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/83/947483/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/83/947483/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/83/947483/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/83/947483/1"}}},"commit":{"parents":[{"commit":"476a14dc0f270784c095c34b789994a1805ae872","subject":"Merge \"Replace deprecated FormatChecker.cls_checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/476a14dc0f270784c095c34b789994a1805ae872"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-16 13:31:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-16 13:31:15.000000000","tz":540},"subject":"Replace UPPER_CONSTRAINTS_FILE","message":"Replace UPPER_CONSTRAINTS_FILE\n\nIt was deprecated in favor of TOX_CONSTRAINTS_FILE.\n\nChange-Id: Icce0fbb080f5b4d5022be8eb82251d916d544fdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/20e3f52bbcdc2fecef0bd6d88bc68ab26a27098b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/20e3f52bbcdc2fecef0bd6d88bc68ab26a27098b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
