)]}'
{"id":"openstack%2Fplacement~957389","triplet_id":"openstack%2Fplacement~master~I518b8073ca41bf0a691a267724be7a94a82b0fb0","project":"openstack/placement","branch":"master","topic":"release-health-check-flamingo","hashtags":[],"change_id":"I518b8073ca41bf0a691a267724be7a94a82b0fb0","subject":"DNM: gate health test","status":"ABANDONED","created":"2025-08-14 09:48:28.000000000","updated":"2025-08-14 17:04:56.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b61fa9f8504350166b679423971d17ccd04926ab","_number":957389,"virtual_id_number":957389,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-08-14 16:58:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-14 16:58:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"10a3fdfa803d77339de774ca51c079c0c9c428ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-08-14 09:48:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0869c00e381d936386dffe17c55862848ee5ed47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 16:58:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/582c643232584a78973b270f6401fd47\n\n- grenade https://zuul.opendev.org/t/openstack/build/1431f8064d53496f98367224373649ae : SUCCESS in 51m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/24c0e81755e04df8a88b8ad10166a08a : SUCCESS in 59m 12s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/7505a63aca3840c5a12a73a9c2a5a28c : SUCCESS in 1h 29m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/934ce236f2544830848b5859a61caf0a : SUCCESS in 32m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9f2e5ea8914426fa8c1bef38998a433 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1422ae30f42548feafb65f22a57cb6ef : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/188f8e4cb7fe45eebfeb150bd134bf36 : SUCCESS in 2m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6e8c952e2b4c4e71b7caa9eb5dfb20ef : SUCCESS in 3m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6f658b4f3f394474bb42aeaa0f91d93b : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beb6119a81974e62a46b077614737d93 : SUCCESS in 10m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1ce4f03784f244dd90a88674b75fa9eb : SUCCESS in 10m 06s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/103a051049294ffa8ea01af1b6b3be33 : SUCCESS in 9m 40s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/add9faed87344c0782b3b60059b9fbd9 : SUCCESS in 20m 42s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/3ecdf72744ad45ebb98c1645f97dbb47 : SUCCESS in 26m 10s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/08ec062fc02a4aba8ee13f41f3d181e0 : FAILURE in 3m 13s (non-voting)\n- tempest-integrated-placement-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e37b423ca854f50b413864bbde4008b : SUCCESS in 1h 51m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aeca0eb58e4a4f0393251858e61970ca : SUCCESS in 47m 02s","accounts_in_message":[],"_revision_number":1},{"id":"b61fa9f8504350166b679423971d17ccd04926ab","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-08-14 17:04:56.000000000","message":"Abandoned\n\ngate jobs are passing, patch can be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ea8d25f6c604c8a2800439104e2de99d85aae1ad","revisions":{"ea8d25f6c604c8a2800439104e2de99d85aae1ad":{"kind":"REWORK","_number":1,"created":"2025-08-14 09:48:28.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/89/957389/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/89/957389/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/89/957389/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/89/957389/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/89/957389/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/89/957389/1"}}},"commit":{"parents":[{"commit":"3e88156836f30b58abb83acf37405788c2ec519b","subject":"Replace deprecated assertItemsEqual","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/3e88156836f30b58abb83acf37405788c2ec519b"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-08-14 09:48:18.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-08-14 09:48:27.000000000","tz":120},"subject":"DNM: gate health test","message":"DNM: gate health test\n\nChange-Id: I518b8073ca41bf0a691a267724be7a94a82b0fb0\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ea8d25f6c604c8a2800439104e2de99d85aae1ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ea8d25f6c604c8a2800439104e2de99d85aae1ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
