)]}'
{"id":"openstack%2Fplacement~958820","triplet_id":"openstack%2Fplacement~master~Iccb8e65bc279d88cee30e30257441bad6d17bde5","project":"openstack/placement","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-08-29 11:47:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iccb8e65bc279d88cee30e30257441bad6d17bde5","subject":"Remove the job with Ubuntu 22.04","status":"MERGED","created":"2025-08-29 01:40:12.000000000","updated":"2025-08-29 11:52:58.000000000","submitted":"2025-08-29 11:47:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958820","meta_rev_id":"4ce39395390b8088342adaa6feb6b866ad2e3d8f","_number":958820,"virtual_id_number":958820,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-29 11:47:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-29 10:08:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2025-08-29 10:08:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2025-08-29 10:08:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-29 02:43:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-29 10:08:00.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2025-08-29 10:08:44.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"}],"messages":[{"id":"d431e42c53ea7b62fc057394173ad70cd39714a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-29 01:40:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ba3a7645b8a68f3ec31be0c2f9e8cbbd46e06bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 02:43:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae56c0b751f046ee84874710cca06cd9\n\n- grenade https://zuul.opendev.org/t/openstack/build/21f53ffa12574251993583bdd7013878 : SUCCESS in 27m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ba6bacebb5b044a2aa4a631dfbcb5263 : SUCCESS in 34m 37s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/ec5bfece4b4b430c99ee90b1c76aa406 : SUCCESS in 53m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a09f929ea9c84bb8809e7f75dc0830a3 : SUCCESS in 56m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19f3a5a039a74a899896096fcb33c737 : SUCCESS in 14m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1541a8dc399d4715930ff0c4725b5fe3 : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e8e72ea9852469aac984860a86e030d : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a4af7cab2c814c82b9a2aa7de91789a9 : SUCCESS in 6m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6a5ca06535614b948425f884e8b8b379 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5080e656cf54af48887b419bec31a00 : SUCCESS in 8m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4dc0ca60d6b543c48cb6999b2f54324d : FAILURE in 4m 27s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/cabaf1d043f24a60a1ae0811a736720b : SUCCESS in 8m 35s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fae656b17d044cf5a97b8b78447d5bb1 : SUCCESS in 24m 34s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/09c8af641dbb403f9e5bb58973e75613 : SUCCESS in 25m 28s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/901cd922edcf4748b39cb7d40557644b : FAILURE in 3m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8c0ce811e03d4d6a8b0f922dc05847c3 : SUCCESS in 57m 18s","accounts_in_message":[],"_revision_number":1},{"id":"2d16e9800050ca69e223b0a86b0c4d281ee091ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-29 03:02:52.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8699f9271b2cbf2508c9ac4991a3c6012c4cdcd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 04:06:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d938555920e0421c917117147d2dc460\n\n- grenade https://zuul.opendev.org/t/openstack/build/3645b2874d4449f3a9aa9032bf2b9e38 : SUCCESS in 1h 00m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9dbed58a54f44c17a17255c0126f44a8 : SUCCESS in 1h 02m 13s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/982c1789536e477fa957e82d4c814d70 : SUCCESS in 54m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df590a6127d94f4184956ff2e53d7a61 : SUCCESS in 32m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0dd2cb181c346aab61891430ba82df2 : SUCCESS in 12m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/199f6c27b7114da687cfc8bb305c6992 : SUCCESS in 6m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/987925c8b66f4927ac2fba67d9edfe7f : SUCCESS in 6m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ed7fdbb3e8354a2fa035a9cdb2c26d70 : SUCCESS in 6m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f34e9a70ab2f4577946dfb5cff9c2eb9 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2b98361b7d14f84847ba9eb4d896cd8 : SUCCESS in 10m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/82a105c0f09f4f9eb5d5c44f3a829237 : SUCCESS in 6m 20s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/64113285943c4cb28179a0dc623eed5f : SUCCESS in 6m 00s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2147658532544711aa0270b6e3ab1ee7 : SUCCESS in 20m 52s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/2c07ba031c77441e81941edc90a75096 : SUCCESS in 27m 39s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/87401a6e800f4efe8f78c6ee1b17a1a2 : FAILURE in 1m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/935aa03f557e44879df134aad12ccbf9 : SUCCESS in 28m 41s","accounts_in_message":[],"_revision_number":2},{"id":"2865b2c981679c325c8470b8a865cc9ddd982a2c","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-08-29 10:08:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9e44c3bfe89d81c61622eb5e7bc10e0a25757ac7","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2025-08-29 10:08:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9a0e6438bbbd469d5a60da645652ca3fa14bfdcb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 10:10:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a1bdc1a1f0d72a1fd921d9274ef32f9916b2bdd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 11:47:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecc3c43bd8744361adfda864d68857e0\n\n- grenade https://zuul.opendev.org/t/openstack/build/8800c6d131d64efaa27a485461303c70 : SUCCESS in 59m 23s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/af33dd6eba194047b3667c78ffa363a0 : SUCCESS in 1h 36m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/441def5ee7374a7db98eabac79903bc0 : SUCCESS in 1h 06m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e33a731142942d8926ff7e6d65cbbbf : SUCCESS in 59m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88052fbe8bda4b56822e44109c8a1bf8 : SUCCESS in 7m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f5e90a7ed806408c9e6121dddb361bea : SUCCESS in 6m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/63451cd0f38748059b78d216405a702b : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6559c37349a54af3a229cd7b3fdb7edf : SUCCESS in 11m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c68e09ef19754e09adf51c866f83b85b : SUCCESS in 8m 17s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d588952b33fe47eabf122ceea04a696d : SUCCESS in 11m 31s\n- placement-nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/02d855d3013d4f268b16487c486dc6f6 : SUCCESS in 24m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a389d9cb40645619cb4d58b59340510 : SUCCESS in 58m 35s","accounts_in_message":[],"_revision_number":2},{"id":"0f821c5333185727aca4687c2659c0d886a6cf6f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 11:47:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4ce39395390b8088342adaa6feb6b866ad2e3d8f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 11:52:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e8b44e3e3ea423f8b2ba2c6a7a14d9f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/632d4824782d4e848bfa11dee260f9f9 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1628b6c55617e3dc2efb1b48f8bd32fa5f39ff59","revisions":{"2ab9bc3cfa54ae4c1ee7f90086ab31586a177b0f":{"kind":"REWORK","_number":1,"created":"2025-08-29 01:40:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/958820/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/20/958820/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/20/958820/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/20/958820/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/20/958820/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/20/958820/1"}}},"commit":{"parents":[{"commit":"3e88156836f30b58abb83acf37405788c2ec519b","subject":"Replace deprecated assertItemsEqual","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/3e88156836f30b58abb83acf37405788c2ec519b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-29 01:38:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-29 01:39:47.000000000","tz":540},"subject":"Remove the job with Ubuntu 22.04","message":"Remove the job with Ubuntu 22.04\n\nUbuntu 22.04 support was removed and now Ubuntu 24.04 is the only\nsupported Ubuntu release.\n\nDrop the job during the transition from 22.04 to 24.04 .\n\nChange-Id: Iccb8e65bc279d88cee30e30257441bad6d17bde5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/2ab9bc3cfa54ae4c1ee7f90086ab31586a177b0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/2ab9bc3cfa54ae4c1ee7f90086ab31586a177b0f"}]},"branch":"refs/heads/master"},"1628b6c55617e3dc2efb1b48f8bd32fa5f39ff59":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-08-29 03:02:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/958820/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/20/958820/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/20/958820/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/20/958820/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/20/958820/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/20/958820/2"}}},"commit":{"parents":[{"commit":"1e32f2babb6b3e11bec992a0a160d0b7df0c82e0","subject":"Remove Python 3.9 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/1e32f2babb6b3e11bec992a0a160d0b7df0c82e0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-29 01:38:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-29 03:02:52.000000000","tz":0},"subject":"Remove the job with Ubuntu 22.04","message":"Remove the job with Ubuntu 22.04\n\nUbuntu 22.04 support was removed and now Ubuntu 24.04 is the only\nsupported Ubuntu release.\n\nDrop the job during the transition from 22.04 to 24.04 .\n\nChange-Id: Iccb8e65bc279d88cee30e30257441bad6d17bde5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/1628b6c55617e3dc2efb1b48f8bd32fa5f39ff59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/1628b6c55617e3dc2efb1b48f8bd32fa5f39ff59"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
