)]}'
{"id":"openstack%2Fplacement~963275","triplet_id":"openstack%2Fplacement~master~I4a43132f5a355d7bf7bfd8caf7b96ec9f62e4b33","project":"openstack/placement","branch":"master","topic":"bug/2126751","attention_set":{},"removed_from_attention_set":{"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2025-10-09 14:33:58.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4a43132f5a355d7bf7bfd8caf7b96ec9f62e4b33","subject":"Release notes for bug/2126751","status":"ABANDONED","created":"2025-10-07 13:01:01.000000000","updated":"2025-10-09 14:33:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a29a9dcf3114d3612af5a0b188059b73473e3c12","_number":963275,"virtual_id_number":963275,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-10-08 14:24:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-07 13:15:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"12eac7b81bbbd1f298beecc698fbc8ce768e7e9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-10-07 13:01:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c15c035f6419bcb8181b5599e83627b4468a492b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 13:15:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/509259b054ac4871bb68088190ba9447\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0916c7a8ba9f4c1e8d38df7d2ca975fa : FAILURE in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ffe94f59ea544b9aa927e5a93f6b304 : SUCCESS in 9m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd49b60027dd4bc69cf9e6829b179231 : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":1},{"id":"f2306d16fe056e04995a4dcb6b7d67c1b0c184d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-10-08 12:29:17.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"30cf1e6418346a45ce19f41b9a2174df33f42908","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 12:45:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f0d46145f094f1ab7e1fda6d983adee\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26f3347c64fb4ff08f67b640a8722e10 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58d7a6e887eb48d1b3efddddf41a0c37 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55b58bf7a1b943599ffb5830eb2e449f : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":2},{"id":"81074e50da13a9d8707e7de83f25c8092abfbb30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-10-08 12:52:24.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"69c35e480baf09ce1e6c9ca5a0e217f8b1b64ee2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-10-08 12:56:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6f10b2752629ce6d6c351d0b8e5be060e3d904a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 13:07:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c917e6ec376940b18e3aac76e3af8749\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc48903f58f04119a3343ab9af1acf29 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c78e891460b443d1ba3f8214b1729ba6 : SUCCESS in 10m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9554bdd6a7224b02b7ca70f02df7ff70 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":4},{"id":"cc219b11b57a14f34238fd8ea5bdffb99cc9dcee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-10-08 13:26:48.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"fd17b5c135be5a2d24b94efe8f75e858f7f7514d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 13:40:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb6988da48be441aaa3d261c4431319c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3446381590c54812a46499ee89625136 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e34249332f149dd90be6872fadc5b63 : SUCCESS in 8m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d161a6eb4a34d208f7414b462d6f542 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":5},{"id":"715217898cb3acabfa04f903b3d0793fe5ad1a78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-10-08 14:10:05.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"ecfa25f79cd760fa3e7776830994983cf5e233a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 14:24:49.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1226c4a48f44354bd21ef7ddc403e2f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca3fa1f3ee2145df93eb3c051988d538 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e56318917a474687b1cc004fd08f8dd2 : SUCCESS in 9m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f49839c2f5aa49578874daf23983871d : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":6},{"id":"a29a9dcf3114d3612af5a0b188059b73473e3c12","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-10-09 14:33:58.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/c/openstack/placement/+/962776/15","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"587a14763bf02cad2b96b8a44f9e74734b44056a","revisions":{"13e1fc3b4db6cc7d5c02940fe45e7b7f4c8bb415":{"kind":"REWORK","_number":1,"created":"2025-10-07 13:01:01.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/75/963275/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/75/963275/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/75/963275/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/75/963275/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/75/963275/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/75/963275/1"}}},"commit":{"parents":[{"commit":"ba237c8a6fa06a8b629bd37cdad0811b48375cb2","subject":"Inline _consolidate_allocation_requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ba237c8a6fa06a8b629bd37cdad0811b48375cb2"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-10-07 12:49:05.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-10-07 12:59:49.000000000","tz":120},"subject":"Release notes for bug/2126751","message":"Release notes for bug/2126751\n\nRelated-Bug: #2126751\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\nChange-Id: I4a43132f5a355d7bf7bfd8caf7b96ec9f62e4b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/13e1fc3b4db6cc7d5c02940fe45e7b7f4c8bb415"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/13e1fc3b4db6cc7d5c02940fe45e7b7f4c8bb415"}]},"branch":"refs/heads/master"},"2c446ecde2079c594fa2be897366666ddd288cf6":{"kind":"REWORK","_number":2,"created":"2025-10-08 12:29:17.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/75/963275/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/75/963275/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/75/963275/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/75/963275/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/75/963275/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/75/963275/2"}}},"commit":{"parents":[{"commit":"cc73376a087c06579a9f9557b055bae903465fe5","subject":"Inline _consolidate_allocation_requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/cc73376a087c06579a9f9557b055bae903465fe5"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-10-07 12:49:05.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-10-08 12:29:07.000000000","tz":120},"subject":"Release notes for bug/2126751","message":"Release notes for bug/2126751\n\nRelated-Bug: #2126751\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\nChange-Id: I4a43132f5a355d7bf7bfd8caf7b96ec9f62e4b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/2c446ecde2079c594fa2be897366666ddd288cf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/2c446ecde2079c594fa2be897366666ddd288cf6"}]},"branch":"refs/heads/master"},"2f7b4bc7aac9321bbd843ba0cb5a4b429c5a448d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-10-08 12:52:24.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/75/963275/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/75/963275/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/75/963275/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/75/963275/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/75/963275/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/75/963275/3"}}},"commit":{"parents":[{"commit":"e46e5440e1a914b1fdb154cb7b3ea67ab70a8ea8","subject":"Inline _consolidate_allocation_requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e46e5440e1a914b1fdb154cb7b3ea67ab70a8ea8"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-10-07 12:49:05.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-10-08 12:52:20.000000000","tz":120},"subject":"Release notes for bug/2126751","message":"Release notes for bug/2126751\n\nRelated-Bug: #2126751\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\nChange-Id: I4a43132f5a355d7bf7bfd8caf7b96ec9f62e4b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/2f7b4bc7aac9321bbd843ba0cb5a4b429c5a448d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/2f7b4bc7aac9321bbd843ba0cb5a4b429c5a448d"}]},"branch":"refs/heads/master"},"e39bf0c0b9796e5c2aa7c552c37b2c9b6a7d2cb1":{"kind":"REWORK","_number":4,"created":"2025-10-08 12:56:51.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/75/963275/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/75/963275/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/75/963275/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/75/963275/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/75/963275/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/75/963275/4"}}},"commit":{"parents":[{"commit":"e46e5440e1a914b1fdb154cb7b3ea67ab70a8ea8","subject":"Inline _consolidate_allocation_requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e46e5440e1a914b1fdb154cb7b3ea67ab70a8ea8"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-10-07 12:49:05.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-10-08 12:56:42.000000000","tz":120},"subject":"Release notes for bug/2126751","message":"Release notes for bug/2126751\n\nRelated-Bug: #2126751\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\nChange-Id: I4a43132f5a355d7bf7bfd8caf7b96ec9f62e4b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e39bf0c0b9796e5c2aa7c552c37b2c9b6a7d2cb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e39bf0c0b9796e5c2aa7c552c37b2c9b6a7d2cb1"}]},"branch":"refs/heads/master"},"3db88075567e4c265919feb439ffb76f2499eb43":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-10-08 13:26:48.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/75/963275/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/75/963275/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/75/963275/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/75/963275/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/75/963275/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/75/963275/5"}}},"commit":{"parents":[{"commit":"da003eadb7ae51b00f705fcffaa0e20a1d5b2d84","subject":"Inline _consolidate_allocation_requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/da003eadb7ae51b00f705fcffaa0e20a1d5b2d84"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-10-07 12:49:05.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-10-08 13:26:41.000000000","tz":120},"subject":"Release notes for bug/2126751","message":"Release notes for bug/2126751\n\nRelated-Bug: #2126751\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\nChange-Id: I4a43132f5a355d7bf7bfd8caf7b96ec9f62e4b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/3db88075567e4c265919feb439ffb76f2499eb43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/3db88075567e4c265919feb439ffb76f2499eb43"}]},"branch":"refs/heads/master"},"587a14763bf02cad2b96b8a44f9e74734b44056a":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-10-08 14:10:05.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/75/963275/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/75/963275/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/75/963275/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/75/963275/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/75/963275/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/75/963275/6"}}},"commit":{"parents":[{"commit":"8077f41e87d4a7e680d0ab579a90b423d5c2edc1","subject":"Inline _consolidate_allocation_requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/8077f41e87d4a7e680d0ab579a90b423d5c2edc1"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-10-07 12:49:05.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-10-08 14:09:59.000000000","tz":120},"subject":"Release notes for bug/2126751","message":"Release notes for bug/2126751\n\nRelated-Bug: #2126751\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\nChange-Id: I4a43132f5a355d7bf7bfd8caf7b96ec9f62e4b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/587a14763bf02cad2b96b8a44f9e74734b44056a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/587a14763bf02cad2b96b8a44f9e74734b44056a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
